)]}'
{"id":"openstack%2Fvenus~889266","triplet_id":"openstack%2Fvenus~master~I1edafaaed9ae04ec599ff02ed9da4bf7835ded9c","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2023-07-25 01:12:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1edafaaed9ae04ec599ff02ed9da4bf7835ded9c","subject":"anomaly detect: add api (get rule)","status":"MERGED","created":"2023-07-24 10:54:35.000000000","updated":"2023-07-25 01:13:56.000000000","submitted":"2023-07-25 01:12:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889266","meta_rev_id":"fdb0944cc392d7f766f12a14a9c9ec80e09eabf8","_number":889266,"virtual_id_number":889266,"owner":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-25 01:12:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":2,"date":"2023-07-25 01:01:19.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":1,"date":"2023-07-25 01:01:19.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-24 11:08:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-25 01:01:19.000000000","updated_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"}],"messages":[{"id":"e265cb8ed7ca7fe4bda015952859b547f79b4004","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-07-24 10:54:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"634e54f6c603382a0ae0a5ec0a7566648055efd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-07-24 10:56:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2f64771286b3876452c5c42350c1e40537deac33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 11:08:07.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1ef245f24a6484295ed75786b070394\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6204ace3773547eb86e120bbcb5dd7b6 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14a1233276ad426b989a5bed13d3bd19 : FAILURE in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aab7b3b4d96c4e558f92d7a2b20f9bf4 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9eec6d5d362420a9d67db9f14f46616 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b89820676ab146578f79d441422bc909 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":2},{"id":"9d6742916b3a8939c238ec55433e0a68f41ec6e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-07-25 00:41:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"52728c4ea9673b62d838b71783499882c919227a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 00:54:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/111af172b84e491eb636ce144913b231\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b836325239964a698718e185c14c0c09 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29988631223b463ea48150c19fcf7ab3 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7d7e08260aa4fbda7d80ad6f09be103 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d68e0882fbb14a7588e0078b9fd8a79f : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd977ea4f89b43078f779ed9ed28d883 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":3},{"id":"b3aad093dc096a1caff0d31198a74ecece43b6ab","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-07-25 01:01:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"14e181bd3624c86536b1ac1bc653a8b96c104ad9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 01:01:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"85fc616e7f5507fe8f3271bc4d83f823777d9906","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 01:12:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/604564c8782f4a34849531ee54699734\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/512c7126774e441598e59d325e6a8f39 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b61d6b3269f449093c1aa355dfe9a9a : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cdad8d1f29f4f75869381a2aa15f5ce : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/813c7e8ebf924597a7e06c88b96b534e : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77caebc7290c40c88a75c514584cc091 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":3},{"id":"a58ec739cb240bb9f63a08988f47c5879aa36f9b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 01:12:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"fdb0944cc392d7f766f12a14a9c9ec80e09eabf8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 01:13:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7856f6082bdf46149f17cf455665ce38\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4575b80e2b164c7d88a386fe08c2043c : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"70d8d02e1b4ca62a8c9c688aed099cdb2e1fb627","revisions":{"05f95555f076ddb29f21ebbd43aa76e309483ce3":{"kind":"REWORK","_number":1,"created":"2023-07-24 10:54:35.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/66/889266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/66/889266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/66/889266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/66/889266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/66/889266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/66/889266/1"}}},"commit":{"parents":[{"commit":"ba0a023d9fb8c398703b1d0c047e51007bceefeb","subject":"anomaly detect: add api (delete rule)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ba0a023d9fb8c398703b1d0c047e51007bceefeb"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-07-24 10:54:24.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-07-24 10:54:24.000000000","tz":0},"subject":"anomaly detect: add api (get rule)","message":"anomaly detect: add api (get rule)\n\nChange-Id: I1edafaaed9ae04ec599ff02ed9da4bf7835ded9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/05f95555f076ddb29f21ebbd43aa76e309483ce3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/05f95555f076ddb29f21ebbd43aa76e309483ce3"}]},"branch":"refs/heads/master"},"1ea97589bccc0fa8d372add1ff8dd224a5fd4e5b":{"kind":"REWORK","_number":2,"created":"2023-07-24 10:56:01.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/66/889266/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/66/889266/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/66/889266/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/66/889266/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/66/889266/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/66/889266/2"}}},"commit":{"parents":[{"commit":"ba0a023d9fb8c398703b1d0c047e51007bceefeb","subject":"anomaly detect: add api (delete rule)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ba0a023d9fb8c398703b1d0c047e51007bceefeb"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-07-24 10:54:24.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-07-24 10:55:49.000000000","tz":0},"subject":"anomaly detect: add api (get rule)","message":"anomaly detect: add api (get rule)\n\nChange-Id: I1edafaaed9ae04ec599ff02ed9da4bf7835ded9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/1ea97589bccc0fa8d372add1ff8dd224a5fd4e5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/1ea97589bccc0fa8d372add1ff8dd224a5fd4e5b"}]},"branch":"refs/heads/master"},"70d8d02e1b4ca62a8c9c688aed099cdb2e1fb627":{"kind":"REWORK","_number":3,"created":"2023-07-25 00:41:04.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/66/889266/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/66/889266/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/66/889266/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/66/889266/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/66/889266/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/66/889266/3"}}},"commit":{"parents":[{"commit":"ba0a023d9fb8c398703b1d0c047e51007bceefeb","subject":"anomaly detect: add api (delete rule)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ba0a023d9fb8c398703b1d0c047e51007bceefeb"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-07-24 10:54:24.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-07-25 00:40:31.000000000","tz":0},"subject":"anomaly detect: add api (get rule)","message":"anomaly detect: add api (get rule)\n\nChange-Id: I1edafaaed9ae04ec599ff02ed9da4bf7835ded9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/70d8d02e1b4ca62a8c9c688aed099cdb2e1fb627"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/70d8d02e1b4ca62a8c9c688aed099cdb2e1fb627"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
