)]}'
{"id":"openstack%2Fvenus~890067","triplet_id":"openstack%2Fvenus~master~Ic9d6c8d99c12728c0acb40d0e1e5a2ef8d13b431","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2023-08-01 05:41:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic9d6c8d99c12728c0acb40d0e1e5a2ef8d13b431","subject":"fix anomaly detect api bugs","status":"MERGED","created":"2023-08-01 03:19:02.000000000","updated":"2023-08-01 05:43:39.000000000","submitted":"2023-08-01 05:41:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890067","meta_rev_id":"dc7411360ed240fd5300ec9c2dab75136e277780","_number":890067,"virtual_id_number":890067,"owner":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-01 05:41:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":2,"date":"2023-08-01 05:32:38.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":1,"date":"2023-08-01 05:32:38.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-01 03:33:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-01 05:32:38.000000000","updated_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"}],"messages":[{"id":"12cd1ba2fc099e2a31052008a9e42e4ab7fd5cfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-08-01 03:19:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27e8b6222d1934bc914ccff5f48807f171b16573","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 03:33:14.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/09a777339a1c469dbd57b8877a860837\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa5c90b86494424d9564e8331d6bfad9 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da92564f19c34d449c25f769b14e6338 : FAILURE in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97d6a7610cbe442eb65f53b6bcdf22ec : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7619adeb01e045fcb5f6114c420d9d06 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55af474192ca4c48bb972f8c95d6f9aa : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a95db4f03a0b930a77b8f4e04aee66ea9e9ca1b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-08-01 04:10:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"42b9d30c4625852295f3a736d6d53a4a58d37d82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-08-01 04:17:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"22df36b4176e96112e353a1a4505e21ccb83fa15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 04:30:02.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/126dfc640b324bf08c389a1b9ee12c54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e3718baa9594db68603baee898e1f03 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56fc0151cfb24c7e8c52cc1b4486782a : FAILURE in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4a4e2d33f494d49ba99338eaf3e259d : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e60fc29e91b14a46a8ab9e341534c4c3 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1001bbcfc0ca454290c7590a51162a09 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":3},{"id":"cbb6aa84626ac064fe787410ab77305f61013eae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-08-01 04:39:00.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"cc96a51f33f45b86108ae3d611520a8e301cb687","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 04:51:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30172aa6d45c4803a5377b4cc09ef88d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf2059f1802b4b5fa89ccb3a5021a8eb : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/947c5432c10a4cdc9295757d2ce6e0e3 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e47d10bc5884f45bca70ca96dfe3168 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/413f67cdabe3411694d3687add100c53 : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81953c850a4a4d529dde3e1b6137b7da : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":4},{"id":"ea5f9901a605315e6eed1af2c0c1e23fe91a302b","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-08-01 05:32:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7fc5497ef07a33548ca69d91bfbe1db3c2b5d356","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 05:32:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a950cc990b4d60334eab6ab40b686380da85c6d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 05:41:48.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13a32d0dba7749ed90cd6a188a53d15d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c7bee3b059448569534bdb17e70d1e8 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b639e10218f04f32ac5ac33c4682efa7 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/793d216c7dea4623b3a261ff1703ac63 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea30f94bd1e64d03a668ac9b8f75205e : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ade22a5cdd094f76b484a5b6820d459b : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":4},{"id":"be3e7770464fab20429b10202430f40b257019ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 05:41:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"dc7411360ed240fd5300ec9c2dab75136e277780","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 05:43:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eafd5d166a664acb8d5040e46e82e72b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4787c2872c2842158aae2963815c1c10 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f92c940b9e3be04b5be7d605e211571d23f26685","revisions":{"83493906e30942d2c8a5af2fe231a907d9523f84":{"kind":"REWORK","_number":1,"created":"2023-08-01 03:19:02.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/67/890067/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/67/890067/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/67/890067/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/67/890067/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/67/890067/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/67/890067/1"}}},"commit":{"parents":[{"commit":"cdf6c050b6f5e30159e850ad8041dd121756faeb","subject":"fix invalid db name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/cdf6c050b6f5e30159e850ad8041dd121756faeb"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-08-01 03:18:49.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-08-01 03:18:49.000000000","tz":0},"subject":"fix anomaly detect api bugs","message":"fix anomaly detect api bugs\n\nChange-Id: Ic9d6c8d99c12728c0acb40d0e1e5a2ef8d13b431\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/83493906e30942d2c8a5af2fe231a907d9523f84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/83493906e30942d2c8a5af2fe231a907d9523f84"}]},"branch":"refs/heads/master"},"e85d7c14a32350b5f2e26dcbc2124adb28fcd982":{"kind":"REWORK","_number":2,"created":"2023-08-01 04:10:44.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/67/890067/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/67/890067/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/67/890067/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/67/890067/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/67/890067/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/67/890067/2"}}},"commit":{"parents":[{"commit":"cdf6c050b6f5e30159e850ad8041dd121756faeb","subject":"fix invalid db name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/cdf6c050b6f5e30159e850ad8041dd121756faeb"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-08-01 03:18:49.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-08-01 04:10:31.000000000","tz":0},"subject":"fix anomaly detect api bugs","message":"fix anomaly detect api bugs\n\nChange-Id: Ic9d6c8d99c12728c0acb40d0e1e5a2ef8d13b431\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/e85d7c14a32350b5f2e26dcbc2124adb28fcd982"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/e85d7c14a32350b5f2e26dcbc2124adb28fcd982"}]},"branch":"refs/heads/master"},"cd8556b2c1914fae16f681596bec01fa7806d1d5":{"kind":"REWORK","_number":3,"created":"2023-08-01 04:17:01.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/67/890067/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/67/890067/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/67/890067/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/67/890067/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/67/890067/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/67/890067/3"}}},"commit":{"parents":[{"commit":"cdf6c050b6f5e30159e850ad8041dd121756faeb","subject":"fix invalid db name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/cdf6c050b6f5e30159e850ad8041dd121756faeb"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-08-01 03:18:49.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-08-01 04:16:51.000000000","tz":0},"subject":"fix anomaly detect api bugs","message":"fix anomaly detect api bugs\n\nChange-Id: Ic9d6c8d99c12728c0acb40d0e1e5a2ef8d13b431\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/cd8556b2c1914fae16f681596bec01fa7806d1d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/cd8556b2c1914fae16f681596bec01fa7806d1d5"}]},"branch":"refs/heads/master"},"f92c940b9e3be04b5be7d605e211571d23f26685":{"kind":"REWORK","_number":4,"created":"2023-08-01 04:39:00.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/67/890067/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/67/890067/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/67/890067/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/67/890067/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/67/890067/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/67/890067/4"}}},"commit":{"parents":[{"commit":"cdf6c050b6f5e30159e850ad8041dd121756faeb","subject":"fix invalid db name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/cdf6c050b6f5e30159e850ad8041dd121756faeb"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-08-01 03:18:49.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-08-01 04:38:46.000000000","tz":0},"subject":"fix anomaly detect api bugs","message":"fix anomaly detect api bugs\n\nChange-Id: Ic9d6c8d99c12728c0acb40d0e1e5a2ef8d13b431\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/f92c940b9e3be04b5be7d605e211571d23f26685"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/f92c940b9e3be04b5be7d605e211571d23f26685"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
