)]}'
{"id":"openstack%2Fvenus~890069","triplet_id":"openstack%2Fvenus~master~Ib6a081276bb3665cd1743d375980c5a9111bc9f2","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2023-08-01 09:02:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib6a081276bb3665cd1743d375980c5a9111bc9f2","subject":"fix detect anomal task bug","status":"MERGED","created":"2023-08-01 08:13:46.000000000","updated":"2023-08-01 09:03:18.000000000","submitted":"2023-08-01 09:02:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890069","meta_rev_id":"51474a7096e1b667d4610c4e663f6757a6ff6954","_number":890069,"virtual_id_number":890069,"owner":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-01 09:02:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":2,"date":"2023-08-01 08:27:31.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":1,"date":"2023-08-01 08:27:31.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-01 08:24:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-01 08:27:31.000000000","updated_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"}],"messages":[{"id":"02373730b5cba62a303900906a69e32989fd4366","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-08-01 08:13:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"451e8eec65fe7426583f46e1f1b8f3db78e8a8f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 08:24:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93d2dab589e049db9639564ad758e0dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae3fff2574994f04858dcf10eac56ac5 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/344b60e8c52148e8b4e0d146276de869 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5295be9859e644a1a5c782c8ec73f847 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95f8f7d919404c5ab57abe20c8e06f78 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95af4bfc08164d7e9aee65ebea6ea6df : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"5e588c0735065aa4f0df0e8292506856c4d6e041","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2023-08-01 08:27:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2bd7ecf1c6f27c7a53ec6ddc4e416ab2d220f8a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 08:27:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"87cda95768d53bd3e4df9f148080ed0b2d17fd83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 09:02:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f48e522f88814e9fa056413fcc8241d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d6d814314534713ba893facdb185210 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b06b2ea7c844e729f3ee9c99c1dc997 : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16fc4211ce384c8a95776da41fa8b1b0 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4920cd2ece2640bd83e09ea9e7b80eb6 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f30d1977cecf49f389503a07e28ae522 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"a1e15d75f0f1c319fe07b860137e9cff68ebcee9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 09:02:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"51474a7096e1b667d4610c4e663f6757a6ff6954","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 09:03:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/584eae26f45e421399549223df4d6592\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a41b49a8a1704b84aad01e511fe44c5c : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d341e44bd768033e3aafe4ee9a3055e55f61ef4","revisions":{"4d341e44bd768033e3aafe4ee9a3055e55f61ef4":{"kind":"REWORK","_number":1,"created":"2023-08-01 08:13:46.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/69/890069/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/69/890069/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/69/890069/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/69/890069/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/69/890069/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/69/890069/1"}}},"commit":{"parents":[{"commit":"22c43afb1c4ec902a6703a44e649ccb3ab11c0be","subject":"rewrite anomaly detect task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/22c43afb1c4ec902a6703a44e649ccb3ab11c0be"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-08-01 08:13:31.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2023-08-01 08:13:31.000000000","tz":0},"subject":"fix detect anomal task bug","message":"fix detect anomal task bug\n\nChange-Id: Ib6a081276bb3665cd1743d375980c5a9111bc9f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/4d341e44bd768033e3aafe4ee9a3055e55f61ef4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/4d341e44bd768033e3aafe4ee9a3055e55f61ef4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
