)]}'
{"id":"openstack%2Fvenus~890769","triplet_id":"openstack%2Fvenus~master~I9e01aa529e41efd49376c7a555c8d431696eff1b","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"33950":{"account":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"last_update":"2023-08-09 07:15:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9e01aa529e41efd49376c7a555c8d431696eff1b","subject":"Fix the sample name in doc","status":"MERGED","created":"2023-08-09 02:49:32.000000000","updated":"2023-08-09 07:15:32.000000000","submitted":"2023-08-09 07:15:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890769","meta_rev_id":"ae1cf44ec301b6ca14e8cb523ad94d8f7b34ecdb","_number":890769,"virtual_id_number":890769,"owner":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-09 07:15:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-09 07:03:15.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-09 07:03:15.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-09 02:56:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-09 07:03:15.000000000","updated_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"}],"messages":[{"id":"4cf37c45921ddd88ffd5d65f8f7908683aebf344","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"date":"2023-08-09 02:49:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89ad60d7c8881b708109bc561ec538e0036ef53d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 02:56:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b815c0d9357471cb44b81987798cbb0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/563909270fb74818aca7c11ef00cb5a0 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bdef654f36b47868c96f556a2829179 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/809f3647cf144c31b6a31363cb44ea8a : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"fe874a5fd1ccc202fd818f1ba86fe852fb9e386b","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-08-09 07:03:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e063cecb34b6fed832daf1f934ee349406fa37ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 07:03:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d13acdfd4a849a79950d4551fe983e294312de5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 07:15:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da2054bf40b74d179d07910e124351f6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e47701f461b1407e9c16d207c9868531 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e07cfd976dd9432f91d1e46871315917 : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a140e60a34a543f7905fd48d1f229bb8 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"ae1cf44ec301b6ca14e8cb523ad94d8f7b34ecdb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 07:15:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2bb867a393086f53614dac13a4e25131509c0ef6","revisions":{"2bb867a393086f53614dac13a4e25131509c0ef6":{"kind":"REWORK","_number":1,"created":"2023-08-09 02:49:32.000000000","uploader":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"ref":"refs/changes/69/890769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/69/890769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/69/890769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/69/890769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/69/890769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/69/890769/1"}}},"commit":{"parents":[{"commit":"5ad46fde3627eaa33d03a8fa56b630d0b3a27d0f","subject":"Merge \"Fix the project name in doc.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/5ad46fde3627eaa33d03a8fa56b630d0b3a27d0f"}]}],"author":{"name":"maliangyi","email":"maliangyi@inspur.com","date":"2023-08-09 02:46:25.000000000","tz":480},"committer":{"name":"maliangyi","email":"maliangyi@inspur.com","date":"2023-08-09 02:49:18.000000000","tz":480},"subject":"Fix the sample name in doc","message":"Fix the sample name in doc\n\nChange-Id: I9e01aa529e41efd49376c7a555c8d431696eff1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/2bb867a393086f53614dac13a4e25131509c0ef6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/2bb867a393086f53614dac13a4e25131509c0ef6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
