)]}'
{"id":"openstack%2Fvenus~890775","triplet_id":"openstack%2Fvenus~master~Ia0266e6ada82267cf1dfe937e40f483a10c6aec8","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2023-08-10 06:16:28.000000000","reason":"\u003cGERRIT_ACCOUNT_27565\u003e replied on the change","reason_account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},"33950":{"account":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"last_update":"2023-08-10 06:26:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia0266e6ada82267cf1dfe937e40f483a10c6aec8","subject":"improve doc format","status":"MERGED","created":"2023-08-10 03:35:41.000000000","updated":"2023-08-10 06:27:35.000000000","submitted":"2023-08-10 06:26:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890775","meta_rev_id":"74ce4fb9815318caf70f59db4f4e415e0b5772ee","_number":890775,"virtual_id_number":890775,"owner":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-10 06:26:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-10 06:16:28.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-10 06:16:28.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-10 03:48:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-10 05:55:02.000000000","updated_by":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"}],"messages":[{"id":"f954e153f68f85459e6a78375cdb12d525dcde5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"date":"2023-08-10 03:35:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35c05334173f6f15530f0e6f4ca3e882a7b2df36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 03:48:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afe8e63db3304c169dc82ac164d323cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/147bccc8c00447cf8b78a38268c9bbab : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3efb999bb6d4482086bdb0cae48fc1b8 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4113f166c00494ab565d09ed9641b7e : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4bf8e7247fe74ec9bf9057531971d8f6 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7abc5cb00c9743efb9587bdb52b40a6d : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"56a43e2ff1298e706f506a54b80f518410c8ecd7","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-08-10 06:16:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c75669cefddf0280875594ac829e4a732200d244","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 06:16:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d67aeac23ff339bdbae864e95812fe70bb3e956","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 06:26:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/120e56dc259d48a8b89104ef1693684b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/807b945683fa4cb383f309a0e532854e : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d6f2766b17d4f57bce50470652233f2 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ecc2bcecde94c27803afbdb2c546381 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02142ad017d44b5ab58b625689eccbf5 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c00c64eb8204f3aacdc1d7ebb002e8c : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"615f4a6dcf1f0617109edb81dec744606ec563fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 06:26:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"74ce4fb9815318caf70f59db4f4e415e0b5772ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 06:27:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb9580a1c95f4c219970c20ea72ec999\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06ebc6f352e64bdfa836844ebcdcda3f : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78d2cf8a8215f1afbaf2c09f8819a0674ec865c4","revisions":{"78d2cf8a8215f1afbaf2c09f8819a0674ec865c4":{"kind":"REWORK","_number":1,"created":"2023-08-10 03:35:41.000000000","uploader":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"ref":"refs/changes/75/890775/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/75/890775/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/75/890775/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/75/890775/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/75/890775/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/75/890775/1"}}},"commit":{"parents":[{"commit":"456a9507935cdc416d16befbf88d9464814d3b56","subject":"Merge \"add request parameters of get anomaly rule list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/456a9507935cdc416d16befbf88d9464814d3b56"}]}],"author":{"name":"maliangyi","email":"maliangyi@inspur.com","date":"2023-08-10 03:15:49.000000000","tz":480},"committer":{"name":"maliangyi","email":"maliangyi@inspur.com","date":"2023-08-10 03:32:24.000000000","tz":480},"subject":"improve doc format","message":"improve doc format\n\nChange-Id: Ia0266e6ada82267cf1dfe937e40f483a10c6aec8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/78d2cf8a8215f1afbaf2c09f8819a0674ec865c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/78d2cf8a8215f1afbaf2c09f8819a0674ec865c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
