)]}'
{"id":"openstack%2Fvenus~890778","triplet_id":"openstack%2Fvenus~master~I4d113b8ea95b1a5191b5bbd9f1f0dcc36dc9bcfc","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"34665":{"account":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"last_update":"2023-08-14 00:10:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4d113b8ea95b1a5191b5bbd9f1f0dcc36dc9bcfc","subject":"Fix missing oslo.versionedobjects library option","status":"MERGED","created":"2023-08-11 05:52:35.000000000","updated":"2023-08-14 00:11:30.000000000","submitted":"2023-08-14 00:10:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890778","meta_rev_id":"518b2d5f8e4f4ead75e2b4feb6bcdf924544a770","_number":890778,"virtual_id_number":890778,"owner":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-14 00:10:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-14 00:06:25.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-14 00:06:25.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-11 06:12:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-14 00:06:25.000000000","updated_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"}],"messages":[{"id":"75ac4b17cbdc7b1bb4191eeacb11e0bf2302c142","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"date":"2023-08-11 05:52:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18269421fbd61aed390a483b8c1aae3ff7d75921","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 06:12:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a85a23b8d10b4a1f8ac89ea8d3dcc888\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba77855294ce4062b814beb6372da465 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d03957d4d744dd58fab1baca97057cb : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbf3daf36eb54766b0ff5cdf915ca0e2 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/529ba1fbbccc47289cc766935cb6f8d3 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8027df5100d94a53a6a5f1156656185e : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":1},{"id":"1dd4188648a0809e34e958ef3790fe0101bacd2d","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-08-14 00:06:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6c4562a6b5c856efdce1109f8bc2dd9f218ce5fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 00:06:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d57e7f3f5ac826d434a9f8c7425bf52ae1cf8e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 00:10:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b379686b634b4adda64e64750e26ea69\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26ada11ff2ec46649fac3b6bb92359b3 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/239b82607fe94d57afabf2c6111f62d8 : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46329285f89343d09277b392f5c06d5d : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/960cf070409c4fe08155860c6d03808f : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31f9746b162847799ab66f0bf98a856f : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8585665d914e99de7bee0b4552bb5eee217169de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 00:10:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"518b2d5f8e4f4ead75e2b4feb6bcdf924544a770","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 00:11:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab4c323d3a134af687494fc7d5620ef3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fe343c500b7434497509bde1be95ce1 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"171d2dae085c40266b0943aec7d4a06c9aae3b76","revisions":{"171d2dae085c40266b0943aec7d4a06c9aae3b76":{"kind":"REWORK","_number":1,"created":"2023-08-11 05:52:35.000000000","uploader":{"_account_id":34665,"name":"anguoming","email":"agm_daydayup@foxmail.com","username":"guomingan"},"ref":"refs/changes/78/890778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/78/890778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/78/890778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/78/890778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/78/890778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/78/890778/1"}}},"commit":{"parents":[{"commit":"542d21b7abfc694358e7b23ea0e11f1745eea967","subject":"delete duplicate content Get anomaly rule list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/542d21b7abfc694358e7b23ea0e11f1745eea967"}]}],"author":{"name":"anguoming","email":"agm_daydayup@foxmail.com","date":"2023-08-11 05:51:55.000000000","tz":480},"committer":{"name":"anguoming","email":"agm_daydayup@foxmail.com","date":"2023-08-11 05:51:55.000000000","tz":480},"subject":"Fix missing oslo.versionedobjects library option","message":"Fix missing oslo.versionedobjects library option\n\nThis ensures the options for oslo.versionedobjects library are\nincluded in the file generated by oslo-config-generator.\n\nChange-Id: I4d113b8ea95b1a5191b5bbd9f1f0dcc36dc9bcfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/171d2dae085c40266b0943aec7d4a06c9aae3b76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/171d2dae085c40266b0943aec7d4a06c9aae3b76"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
