)]}'
{"id":"openstack%2Fvenus~890866","triplet_id":"openstack%2Fvenus~master~Ia19c6a95234df97af7e88de13e78af71a92ad530","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"33436":{"account":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"last_update":"2023-08-09 07:15:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia19c6a95234df97af7e88de13e78af71a92ad530","subject":"Fix the project name in tox.ini.","status":"MERGED","created":"2023-08-09 01:29:38.000000000","updated":"2023-08-09 07:16:37.000000000","submitted":"2023-08-09 07:15:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890866","meta_rev_id":"218b2df9e6fd7ecea070c8809c91261bed102fc2","_number":890866,"virtual_id_number":890866,"owner":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23360,"name":"Eric Lei","email":"leiyashuai@inspur.com","username":"leiyashuai"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-09 07:15:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"recommended":{"_account_id":23360,"name":"Eric Lei","email":"leiyashuai@inspur.com","username":"leiyashuai"},"all":[{"value":1,"date":"2023-08-09 01:36:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23360,"name":"Eric Lei","email":"leiyashuai@inspur.com","username":"leiyashuai"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-09 07:02:37.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":23360,"name":"Eric Lei","email":"leiyashuai@inspur.com","username":"leiyashuai"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-09 07:02:37.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23360,"name":"Eric Lei","email":"leiyashuai@inspur.com","username":"leiyashuai"},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-09 01:36:33.000000000","updated_by":{"_account_id":23360,"name":"Eric Lei","email":"leiyashuai@inspur.com","username":"leiyashuai"},"reviewer":{"_account_id":23360,"name":"Eric Lei","email":"leiyashuai@inspur.com","username":"leiyashuai"},"state":"REVIEWER"},{"updated":"2023-08-09 01:42:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-09 07:02:37.000000000","updated_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"}],"messages":[{"id":"a3ccf9b9c6679798973419eb9033ffbc070389c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"date":"2023-08-09 01:29:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc20e626926a93160937802c6c020fb11820ab28","author":{"_account_id":23360,"name":"Eric Lei","email":"leiyashuai@inspur.com","username":"leiyashuai"},"date":"2023-08-09 01:36:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c08fe7d513e783282f6c22bc5a80804e625dd474","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 01:42:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/949587962afc49d29bc4f09acc57d63a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/acb6576e6f224ef28806294b818e10c9 : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e3287fdb9b5482b844aa63d51b061cf : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac5c7ee0d6904867b2ff2e07a872f5cb : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5dd4df33c3848a098f91c7a813ca6a9 : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb19241b73384e71880f64e3ff1c000d : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6887334fc9248fd82979a276591951c : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b188975e719a48cca7ada72c3c7ef798 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"8d9113abfede8c84119131d2cfc25124b1a5fd91","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-08-09 07:02:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c5da38289832f1f6da1b858d053ebb593dcaaeec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 07:02:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93f55d0d41df2315fc93d6ef3cf7c0a7f8b36b9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 07:15:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f73927549444f959ea164cf02d29d5d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e59d96edac6a4479aab92f0442518ced : SUCCESS in 4m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d66ef24e27f6440b8cc83c693a11c26e : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a774db3ffd744618b9a8c14a603e67e8 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/757b958669d843e09a4355570a570d99 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36e37af21ae34bbba197a7c9cb1da5ca : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64816bf50b784d298443249c46498d61 : SUCCESS in 7m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b2fa4339c55463d889858483d097ee3 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"792c6bde1773565dd80886746ef99dd8bd218249","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 07:15:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"218b2df9e6fd7ecea070c8809c91261bed102fc2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 07:16:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/265b1068ec4a4d84b587e9cf0f6415b7\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/7adbc97e2079489eac177101431847ca : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5ac95b28e074145abe0ee2609791136 : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/423407695969423bb982b53f54909279 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"497b4adeb7c44c859f8049ddcabcdfa9481431dd","revisions":{"497b4adeb7c44c859f8049ddcabcdfa9481431dd":{"kind":"REWORK","_number":1,"created":"2023-08-09 01:29:38.000000000","uploader":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"ref":"refs/changes/66/890866/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/66/890866/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/66/890866/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/66/890866/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/66/890866/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/66/890866/1"}}},"commit":{"parents":[{"commit":"5ad46fde3627eaa33d03a8fa56b630d0b3a27d0f","subject":"Merge \"Fix the project name in doc.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/5ad46fde3627eaa33d03a8fa56b630d0b3a27d0f"}]}],"author":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2023-08-09 01:28:48.000000000","tz":480},"committer":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2023-08-09 01:28:57.000000000","tz":480},"subject":"Fix the project name in tox.ini.","message":"Fix the project name in tox.ini.\n\nThe project name is wrong, fix it.\n\nChange-Id: Ia19c6a95234df97af7e88de13e78af71a92ad530\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/497b4adeb7c44c859f8049ddcabcdfa9481431dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/497b4adeb7c44c859f8049ddcabcdfa9481431dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
