)]}'
{"id":"openstack%2Fvenus~890881","triplet_id":"openstack%2Fvenus~master~Ie702712e81fbdaaeee36bc8c377074904e33307a","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"36229":{"account":{"_account_id":36229,"name":"Natz cheng","display_name":"Natzcheng","email":"chenggang01@inspur.com","username":"Natzcheng"},"last_update":"2023-08-09 10:45:56.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie702712e81fbdaaeee36bc8c377074904e33307a","subject":"add api doc: add sample of get anomaly rule list","status":"ABANDONED","created":"2023-08-09 07:09:48.000000000","updated":"2023-08-09 10:45:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"19c4ec6c3f59df9394a30c1c61dfb6f4d1954255","_number":890881,"virtual_id_number":890881,"owner":{"_account_id":36229,"name":"Natz cheng","display_name":"Natzcheng","email":"chenggang01@inspur.com","username":"Natzcheng"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-09 07:18:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46ef66ceb141e5168a3d237a43e2241c8e3f7891","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36229,"name":"Natz cheng","display_name":"Natzcheng","email":"chenggang01@inspur.com","username":"Natzcheng"},"date":"2023-08-09 07:09:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2306b1b6adfdb8a313f52bb70a9f39c8d8a9f2b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 07:18:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef44954302fc4bebae2afda8a0dc0eb8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/43211894b37944fb8f09763592156248 : FAILURE in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13c9aea6b64346809e869c1984efd814 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29203b2dabef41f48d9adb732083669f : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"24b89acbd68c1cf761fc85aa7c3d4588c8a639a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36229,"name":"Natz cheng","display_name":"Natzcheng","email":"chenggang01@inspur.com","username":"Natzcheng"},"date":"2023-08-09 08:12:14.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"19c4ec6c3f59df9394a30c1c61dfb6f4d1954255","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36229,"name":"Natz cheng","display_name":"Natzcheng","email":"chenggang01@inspur.com","username":"Natzcheng"},"date":"2023-08-09 10:45:56.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"11ac9b7200340318f0ec176b62dbf94981f37393","revisions":{"e43e39523207fdbc9cbf3f6ee03f1af214c5a37f":{"kind":"REWORK","_number":1,"created":"2023-08-09 07:09:48.000000000","uploader":{"_account_id":36229,"name":"Natz cheng","display_name":"Natzcheng","email":"chenggang01@inspur.com","username":"Natzcheng"},"ref":"refs/changes/81/890881/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/81/890881/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/81/890881/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/81/890881/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/81/890881/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/81/890881/1"}}},"commit":{"parents":[{"commit":"c77b24d5d8f93e551f08ae4ffbe701c157a230a2","subject":"add api doc: add sample of get anomaly rule list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/c77b24d5d8f93e551f08ae4ffbe701c157a230a2"}]}],"author":{"name":"Natzcheng","email":"chenggang01@inspur.com","date":"2023-08-09 07:09:11.000000000","tz":480},"committer":{"name":"Natzcheng","email":"chenggang01@inspur.com","date":"2023-08-09 07:09:11.000000000","tz":480},"subject":"add api doc: add sample of get anomaly rule list","message":"add api doc: add sample of get anomaly rule list\n\nChange-Id: Ie702712e81fbdaaeee36bc8c377074904e33307a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/e43e39523207fdbc9cbf3f6ee03f1af214c5a37f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/e43e39523207fdbc9cbf3f6ee03f1af214c5a37f"}]},"branch":"refs/heads/master"},"11ac9b7200340318f0ec176b62dbf94981f37393":{"kind":"REWORK","_number":2,"created":"2023-08-09 08:12:14.000000000","uploader":{"_account_id":36229,"name":"Natz cheng","display_name":"Natzcheng","email":"chenggang01@inspur.com","username":"Natzcheng"},"ref":"refs/changes/81/890881/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/81/890881/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/81/890881/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/81/890881/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/81/890881/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/81/890881/2"}}},"commit":{"parents":[{"commit":"c77b24d5d8f93e551f08ae4ffbe701c157a230a2","subject":"add api doc: add sample of get anomaly rule list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/c77b24d5d8f93e551f08ae4ffbe701c157a230a2"}]}],"author":{"name":"Natzcheng","email":"chenggang01@inspur.com","date":"2023-08-09 07:09:11.000000000","tz":480},"committer":{"name":"Natz cheng","email":"chenggang01@inspur.com","date":"2023-08-09 08:12:14.000000000","tz":0},"subject":"add api doc: add sample of get anomaly rule list","message":"add api doc: add sample of get anomaly rule list\n\nChange-Id: Ie702712e81fbdaaeee36bc8c377074904e33307a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/11ac9b7200340318f0ec176b62dbf94981f37393"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/11ac9b7200340318f0ec176b62dbf94981f37393"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
