)]}'
{"id":"openstack%2Fvenus~890903","triplet_id":"openstack%2Fvenus~master~I0031cafe2f3daf1af671c826704ebd96781e305b","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"35751":{"account":{"_account_id":35751,"name":"LiZekun","display_name":"LiZekun","email":"lizekun02@inspur.com","username":"joeylee888888"},"last_update":"2023-08-10 01:53:20.000000000","reason":"Change was submitted"},"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2023-08-10 01:49:07.000000000","reason":"\u003cGERRIT_ACCOUNT_27565\u003e replied on the change","reason_account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},"30623":{"account":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"last_update":"2023-08-10 01:53:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0031cafe2f3daf1af671c826704ebd96781e305b","subject":"add api doc: add sample of delete anomaly rule","status":"MERGED","created":"2023-08-09 11:12:51.000000000","updated":"2023-08-10 01:54:20.000000000","submitted":"2023-08-10 01:53:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890903","meta_rev_id":"ee7d520cc651e85b65054583694aaeaf68167a39","_number":890903,"virtual_id_number":890903,"owner":{"_account_id":35751,"name":"LiZekun","display_name":"LiZekun","email":"lizekun02@inspur.com","username":"joeylee888888"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-10 01:53:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-10 01:49:07.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-10 01:49:07.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-09 11:24:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-09 11:43:37.000000000","updated_by":{"_account_id":35751,"name":"LiZekun","display_name":"LiZekun","email":"lizekun02@inspur.com","username":"joeylee888888"},"reviewer":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"state":"REVIEWER"},{"updated":"2023-08-09 12:00:56.000000000","updated_by":{"_account_id":35751,"name":"LiZekun","display_name":"LiZekun","email":"lizekun02@inspur.com","username":"joeylee888888"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"}],"messages":[{"id":"0136f3f787484efbb7a6f09353b6e5106b4e217c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35751,"name":"LiZekun","display_name":"LiZekun","email":"lizekun02@inspur.com","username":"joeylee888888"},"date":"2023-08-09 11:12:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35082d718c3e855e48e478d9ad4100ef87b6f8f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 11:24:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdde3ada9339456f8e63b97f230adfb5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e2f1c8fb8a9644428d11031c210d92b0 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/017d35e09abf4eb583f6743d12a1000a : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39b78d6f89e44c99ac66f4a6151369c6 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"1257315d85d8ce32f07e6d50b690c821e40464ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35751,"name":"LiZekun","display_name":"LiZekun","email":"lizekun02@inspur.com","username":"joeylee888888"},"date":"2023-08-09 13:37:49.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f7b28247f940a76fde952048d2bea6309f55eb66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 13:49:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4700073a7fbf486ca7a6bb6ab477beb1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a433da85b5744e40a6fe53d4d63470e3 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9a6e7e810ec49309a7e4290cb7af5f9 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aa244a38990411c93a3864041c49f9a : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":2},{"id":"c83daea4f1c0098209d1741bafa4daf48dad5d16","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-08-10 01:49:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"84995ccfb4a11b8adcf6c3f4ba81e1821a1dee0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 01:49:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d0b06ec6484fe2fa59506e15a24796cfd47180ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 01:53:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d69d24ab3ff4ca397ff6ba34e49473b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3890f7e8b64f45148024948f894e4c5c : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02afe9c983cc4f4199f1d752e6c9d55b : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54949313b85c42b1b2070490bf37c3e5 : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":2},{"id":"6f6d00fcebb39a144a0627b16267d253804a7c00","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 01:53:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ee7d520cc651e85b65054583694aaeaf68167a39","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 01:54:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f724a833ee5a43b08fe1a8cc844b0bda\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/02a2eae238ae44148c2f096d1cd3aa5d : SUCCESS in 40s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd9f69f02f0f404088686545e8ffea37 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8af4d2e447e326696c587854a6481267a2b81efa","revisions":{"ff28a3a22b36ad7e23831480b810eb1dfd7c46fc":{"kind":"REWORK","_number":1,"created":"2023-08-09 11:12:51.000000000","uploader":{"_account_id":35751,"name":"LiZekun","display_name":"LiZekun","email":"lizekun02@inspur.com","username":"joeylee888888"},"ref":"refs/changes/03/890903/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/03/890903/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/03/890903/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/03/890903/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/03/890903/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/03/890903/1"}}},"commit":{"parents":[{"commit":"e7926244a16a0177419ee3deb96d87178fd4a892","subject":"Merge \"add api doc: add sample of get anomaly rule list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/e7926244a16a0177419ee3deb96d87178fd4a892"}]}],"author":{"name":"LiZekun","email":"lizekun02@inspur.com","date":"2023-08-09 11:11:30.000000000","tz":480},"committer":{"name":"LiZekun","email":"lizekun02@inspur.com","date":"2023-08-09 11:11:30.000000000","tz":480},"subject":"add api doc: add sample of delete anomaly rule","message":"add api doc: add sample of delete anomaly rule\n\nChange-Id: I0031cafe2f3daf1af671c826704ebd96781e305b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ff28a3a22b36ad7e23831480b810eb1dfd7c46fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ff28a3a22b36ad7e23831480b810eb1dfd7c46fc"}]},"branch":"refs/heads/master"},"8af4d2e447e326696c587854a6481267a2b81efa":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-08-09 13:37:49.000000000","uploader":{"_account_id":35751,"name":"LiZekun","display_name":"LiZekun","email":"lizekun02@inspur.com","username":"joeylee888888"},"ref":"refs/changes/03/890903/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/03/890903/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/03/890903/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/03/890903/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/03/890903/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/03/890903/2"}}},"commit":{"parents":[{"commit":"5172929f9de9ec87adc1ebea1ba226f2dc0ae4bb","subject":"Add api \u0027update anomaly rule\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/5172929f9de9ec87adc1ebea1ba226f2dc0ae4bb"}]}],"author":{"name":"LiZekun","email":"lizekun02@inspur.com","date":"2023-08-09 11:11:30.000000000","tz":480},"committer":{"name":"LiZekun","email":"lizekun02@inspur.com","date":"2023-08-09 13:16:50.000000000","tz":480},"subject":"add api doc: add sample of delete anomaly rule","message":"add api doc: add sample of delete anomaly rule\n\nChange-Id: I0031cafe2f3daf1af671c826704ebd96781e305b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8af4d2e447e326696c587854a6481267a2b81efa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8af4d2e447e326696c587854a6481267a2b81efa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
