)]}'
{"id":"openstack%2Fvenus~890910","triplet_id":"openstack%2Fvenus~master~I8c00be51e07ad9372b279393ab8cfcf166ff23d0","project":"openstack/venus","branch":"master","topic":"update","attention_set":{},"removed_from_attention_set":{"23360":{"account":{"_account_id":23360,"name":"Eric Lei","email":"leiyashuai@inspur.com","username":"leiyashuai"},"last_update":"2023-08-10 01:46:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8c00be51e07ad9372b279393ab8cfcf166ff23d0","subject":"Add request parameters of api \u0027update anomaly rule\u0027","status":"MERGED","created":"2023-08-09 12:12:02.000000000","updated":"2023-08-10 01:47:58.000000000","submitted":"2023-08-10 01:46:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890910-update","meta_rev_id":"37a991cc32f6e5c9f1674a14cdbae3cf82a6a61d","_number":890910,"virtual_id_number":890910,"owner":{"_account_id":23360,"name":"Eric Lei","email":"leiyashuai@inspur.com","username":"leiyashuai"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-10 01:46:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-10 01:07:38.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-10 01:07:38.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-09 12:21:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-10 01:07:38.000000000","updated_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"}],"messages":[{"id":"4f9e0031820d1e7effe79e3febbee5ddd2e1f603","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23360,"name":"Eric Lei","email":"leiyashuai@inspur.com","username":"leiyashuai"},"date":"2023-08-09 12:12:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c7e3cef3de17d2a366fe6ce16ca9db010c3ed9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 12:21:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb474469b6fb4967a6ed8551207072fd\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0bd9212639114dedaa987ddbe4b48689 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fe8c9452d1343f8b271c12dba54f269 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcd2381137354c4c84b83bc2513c2106 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"1f8e709efc8de9b90082a56685bfd74293df2904","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-08-10 01:07:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c533afce851607b2d837d292e72a7ab5bc8ddbef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 01:08:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a559c02b92f94f401e4ce2b69086dfc2ecf4f4de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 01:46:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7da5c85ee54342f0a0a11377f412c658\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6ef74654b013456e98a10e40a62368b1 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a16ac2e497c8463d9fcfd8eb9b52a6cc : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27e875230d6f4fb8838c98227e38bcb8 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"0f23136113cf446fe3d17de481108f2f7aefa030","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 01:46:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"37a991cc32f6e5c9f1674a14cdbae3cf82a6a61d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 01:47:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00988f3d46fc4414b8f99f76c154b9ca\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/feb5ca3599ab4e9d980cc41635a99269 : SUCCESS in 42s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2eb0a58be4554c2ba312ebec007fc765 : SUCCESS in 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f89035688109278ef075285d0fc20dde537672ae","revisions":{"f89035688109278ef075285d0fc20dde537672ae":{"kind":"REWORK","_number":1,"created":"2023-08-09 12:12:02.000000000","uploader":{"_account_id":23360,"name":"Eric Lei","email":"leiyashuai@inspur.com","username":"leiyashuai"},"ref":"refs/changes/10/890910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/10/890910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/10/890910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/10/890910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/10/890910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/10/890910/1"}}},"commit":{"parents":[{"commit":"5172929f9de9ec87adc1ebea1ba226f2dc0ae4bb","subject":"Add api \u0027update anomaly rule\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/5172929f9de9ec87adc1ebea1ba226f2dc0ae4bb"}]}],"author":{"name":"leiyashuai","email":"leiyashuai@inspur.com","date":"2023-08-09 12:11:04.000000000","tz":480},"committer":{"name":"leiyashuai","email":"leiyashuai@inspur.com","date":"2023-08-09 12:11:04.000000000","tz":480},"subject":"Add request parameters of api \u0027update anomaly rule\u0027","message":"Add request parameters of api \u0027update anomaly rule\u0027\n\nChange-Id: I8c00be51e07ad9372b279393ab8cfcf166ff23d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/f89035688109278ef075285d0fc20dde537672ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/f89035688109278ef075285d0fc20dde537672ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
