)]}'
{"id":"openstack%2Fvenus~890915","triplet_id":"openstack%2Fvenus~master~Ia02b25d20216c4b2476e029fff564c335974eed1","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"35751":{"account":{"_account_id":35751,"name":"LiZekun","display_name":"LiZekun","email":"lizekun02@inspur.com","username":"joeylee888888"},"last_update":"2023-08-10 01:57:10.000000000","reason":"Change was submitted"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2023-08-10 01:57:10.000000000","reason":"Change was submitted"},"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2023-08-10 01:48:54.000000000","reason":"\u003cGERRIT_ACCOUNT_27565\u003e replied on the change","reason_account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},"30623":{"account":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"last_update":"2023-08-10 01:57:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia02b25d20216c4b2476e029fff564c335974eed1","subject":"add api doc: delete anomaly detect rule","status":"MERGED","created":"2023-08-09 13:37:49.000000000","updated":"2023-08-10 01:58:07.000000000","submitted":"2023-08-10 01:57:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890915","meta_rev_id":"605bdde39cebf5f80795445c120a53dc26791378","_number":890915,"virtual_id_number":890915,"owner":{"_account_id":35751,"name":"LiZekun","display_name":"LiZekun","email":"lizekun02@inspur.com","username":"joeylee888888"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-10 01:57:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-10 01:48:54.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-10 01:48:54.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-09 13:42:41.000000000","updated_by":{"_account_id":35751,"name":"LiZekun","display_name":"LiZekun","email":"lizekun02@inspur.com","username":"joeylee888888"},"reviewer":{"_account_id":30623,"name":"suzhengwei","email":"suzhengwei@inspur.com","username":"SuZhengwei"},"state":"REVIEWER"},{"updated":"2023-08-09 13:42:41.000000000","updated_by":{"_account_id":35751,"name":"LiZekun","display_name":"LiZekun","email":"lizekun02@inspur.com","username":"joeylee888888"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2023-08-09 13:42:41.000000000","updated_by":{"_account_id":35751,"name":"LiZekun","display_name":"LiZekun","email":"lizekun02@inspur.com","username":"joeylee888888"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"},{"updated":"2023-08-09 13:46:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"de7214463318078f0470e3130b9389487fb7600e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35751,"name":"LiZekun","display_name":"LiZekun","email":"lizekun02@inspur.com","username":"joeylee888888"},"date":"2023-08-09 13:37:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61c6ac182245404699aed068c450a0033097ec84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35751,"name":"LiZekun","display_name":"LiZekun","email":"lizekun02@inspur.com","username":"joeylee888888"},"date":"2023-08-09 13:41:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9936496d83eb90322e5ef558c14979fb3c9b8b29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 13:46:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd66eb67a9054329a9ce21778f8953a1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8162ec34daff4ecd8984401b71e18f66 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43b8dd0752094dd3b83b0261228b03f3 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a1b847a04f048daa94b744fd8b717bc : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":2},{"id":"40fc05afd1e8569f905ac943a87fe7697ca8399e","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-08-10 01:48:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"11617c537846588991c1d6e51044ca8a5be150f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 01:49:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a0bc610a3927570fa6949fc1bdaf34a6c08e4925","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 01:57:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a527633e86d4c5ab35a8ddc865e5ede\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5605c61d465a4b7588db6c2eb97d8a32 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8092fbc5cdec42d594c91b126b7fea20 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4368ccf6bed4ab8bd720c86f6cdaee8 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":2},{"id":"ef5c5e79b240a85037bab901a4f4121dbea2ab51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 01:57:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"605bdde39cebf5f80795445c120a53dc26791378","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 01:58:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5cbe6847ae04322b89412f0207cbf2c\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/1ec95283e4524e8fb1d928b1005ad241 : SUCCESS in 37s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/623be045c11a45fc873fb1f9417ab612 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ec7225934130646f87be139b98e83ece5ee75324","revisions":{"a81b250e05e8e053c508e26bf62047f027ebff2e":{"kind":"REWORK","_number":1,"created":"2023-08-09 13:37:49.000000000","uploader":{"_account_id":35751,"name":"LiZekun","display_name":"LiZekun","email":"lizekun02@inspur.com","username":"joeylee888888"},"ref":"refs/changes/15/890915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/15/890915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/15/890915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/15/890915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/15/890915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/15/890915/1"}}},"commit":{"parents":[{"commit":"8af4d2e447e326696c587854a6481267a2b81efa","subject":"add api doc: add sample of delete anomaly rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8af4d2e447e326696c587854a6481267a2b81efa"}]}],"author":{"name":"LiZekun","email":"lizekun02@inspur.com","date":"2023-08-09 13:15:53.000000000","tz":480},"committer":{"name":"LiZekun","email":"lizekun02@inspur.com","date":"2023-08-09 13:36:54.000000000","tz":480},"subject":"add api doc: delete anomaly detect rule","message":"add api doc: delete anomaly detect rule\n\nChange-Id: Ia02b25d20216c4b2476e029fff564c335974eed1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/a81b250e05e8e053c508e26bf62047f027ebff2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/a81b250e05e8e053c508e26bf62047f027ebff2e"}]},"branch":"refs/heads/master"},"ec7225934130646f87be139b98e83ece5ee75324":{"kind":"REWORK","_number":2,"created":"2023-08-09 13:41:16.000000000","uploader":{"_account_id":35751,"name":"LiZekun","display_name":"LiZekun","email":"lizekun02@inspur.com","username":"joeylee888888"},"ref":"refs/changes/15/890915/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/15/890915/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/15/890915/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/15/890915/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/15/890915/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/15/890915/2"}}},"commit":{"parents":[{"commit":"8af4d2e447e326696c587854a6481267a2b81efa","subject":"add api doc: add sample of delete anomaly rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/8af4d2e447e326696c587854a6481267a2b81efa"}]}],"author":{"name":"LiZekun","email":"lizekun02@inspur.com","date":"2023-08-09 13:15:53.000000000","tz":480},"committer":{"name":"LiZekun","email":"lizekun02@inspur.com","date":"2023-08-09 13:40:53.000000000","tz":480},"subject":"add api doc: delete anomaly detect rule","message":"add api doc: delete anomaly detect rule\n\nChange-Id: Ia02b25d20216c4b2476e029fff564c335974eed1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ec7225934130646f87be139b98e83ece5ee75324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ec7225934130646f87be139b98e83ece5ee75324"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
