)]}'
{"id":"openstack%2Fvenus~891006","triplet_id":"openstack%2Fvenus~master~If1b4d9335e23570bd323f64e418d9bc6b99d5365","project":"openstack/venus","branch":"master","topic":"update","attention_set":{},"removed_from_attention_set":{"23360":{"account":{"_account_id":23360,"name":"Eric Lei","email":"leiyashuai@inspur.com","username":"leiyashuai"},"last_update":"2023-08-10 02:56:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If1b4d9335e23570bd323f64e418d9bc6b99d5365","subject":"Add sample of API \u0027update anomaly rule\u0027","status":"MERGED","created":"2023-08-10 02:34:46.000000000","updated":"2023-08-10 02:57:59.000000000","submitted":"2023-08-10 02:56:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891006-update","meta_rev_id":"47889c65d631f0a90a03f1999f8efca178b058e7","_number":891006,"virtual_id_number":891006,"owner":{"_account_id":23360,"name":"Eric Lei","email":"leiyashuai@inspur.com","username":"leiyashuai"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-10 02:56:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-10 02:46:02.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-10 02:46:02.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-10 02:45:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-10 02:46:02.000000000","updated_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"}],"messages":[{"id":"4429ecea82f8ae55bac01ede8d85c467a60f25de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23360,"name":"Eric Lei","email":"leiyashuai@inspur.com","username":"leiyashuai"},"date":"2023-08-10 02:34:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5677d40cd97115593f2e27c48c94d219dcb33a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 02:45:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f312dbdb740448b89edb06e3bf08956\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ec9243316325457d91dfaadfcc06f294 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae869a0c0789495d8cd58d8929d57521 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d4a75d0e1774b028db31f4c0d22bf9a : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c8efa1dc2b551cb8db094c2c9a9ea108169e0f58","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-08-10 02:46:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ca0d7055ef2095225df990ce966920f51f679224","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 02:46:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e1c8bf8e458beee6d795852833a8b52de14a25d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 02:56:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca0783dece954e0fb18eb77161e70df2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2436fd02e2ec4af4b749c50b85c896e8 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20b767be6ca945b4a9809f2379033f37 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/263cffebff394f69bf2e74314d73a51f : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8bd39e5d081c2561b46f0282adf356a179e14b3a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 02:56:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"47889c65d631f0a90a03f1999f8efca178b058e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 02:57:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d574e938a2544e69b8c2eea1fdb0377\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/7242b2b2c60d47edbb79f79d0f983b62 : SUCCESS in 45s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4d4ce3b498b4af48aa1c3e52ea6ee82 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f04e972195fd94ed7190f9125edc4622f6b3e5c9","revisions":{"f04e972195fd94ed7190f9125edc4622f6b3e5c9":{"kind":"REWORK","_number":1,"created":"2023-08-10 02:34:46.000000000","uploader":{"_account_id":23360,"name":"Eric Lei","email":"leiyashuai@inspur.com","username":"leiyashuai"},"ref":"refs/changes/06/891006/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/06/891006/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/06/891006/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/06/891006/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/06/891006/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/06/891006/1"}}},"commit":{"parents":[{"commit":"cb75ca6ee97af6dbd50a12f6e4d0c4993fd38003","subject":"Merge \"Fix the respose name of records\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/cb75ca6ee97af6dbd50a12f6e4d0c4993fd38003"}]}],"author":{"name":"leiyashuai","email":"leiyashuai@inspur.com","date":"2023-08-10 02:34:18.000000000","tz":480},"committer":{"name":"leiyashuai","email":"leiyashuai@inspur.com","date":"2023-08-10 02:34:18.000000000","tz":480},"subject":"Add sample of API \u0027update anomaly rule\u0027","message":"Add sample of API \u0027update anomaly rule\u0027\n\nChange-Id: If1b4d9335e23570bd323f64e418d9bc6b99d5365\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/f04e972195fd94ed7190f9125edc4622f6b3e5c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/f04e972195fd94ed7190f9125edc4622f6b3e5c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
