)]}'
{"id":"openstack%2Fvenus~891015","triplet_id":"openstack%2Fvenus~master~I46f0637f41bec279166c1fe9eb34ae420e3e715c","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"36229":{"account":{"_account_id":36229,"name":"Natz cheng","display_name":"Natzcheng","email":"chenggang01@inspur.com","username":"Natzcheng"},"last_update":"2023-08-10 08:23:08.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I46f0637f41bec279166c1fe9eb34ae420e3e715c","subject":"support request parameter log_type","status":"ABANDONED","created":"2023-08-10 07:01:27.000000000","updated":"2023-08-10 08:23:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"56113c49fb4aee1f0623f14992f674890af961a0","_number":891015,"virtual_id_number":891015,"owner":{"_account_id":36229,"name":"Natz cheng","display_name":"Natzcheng","email":"chenggang01@inspur.com","username":"Natzcheng"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-10 07:11:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2023-08-10 08:14:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-08-10 08:15:36.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-10 07:11:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-10 08:14:15.000000000","updated_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"}],"messages":[{"id":"307912346b5b1a94133ffab519541f2a2d4537e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36229,"name":"Natz cheng","display_name":"Natzcheng","email":"chenggang01@inspur.com","username":"Natzcheng"},"date":"2023-08-10 07:01:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e49e9f3181d1ddabbb476aaf387cfa83e53c92b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36229,"name":"Natz cheng","display_name":"Natzcheng","email":"chenggang01@inspur.com","username":"Natzcheng"},"date":"2023-08-10 07:02:26.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"187c130308a761c32788d78f070a16143bdb0922","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 07:11:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0590e57809e249b9a4d95bddfdd5c8a9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/88b1c8d321ef43fcb0511cd3130dfadd : FAILURE in 3m 48s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/192d5711014c4faa899e9e0c7f204438 : SUCCESS in 1m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9962393dbd1242189665037e08e71514 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66414792b25a49fca8b7d48807d7dbd6 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0f256eb16f649c9a79addfee23187d7 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80edd725267f4997a8715bc1ac30789d : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a30cb07cd55e45efaa01953ad805c8d5 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fda6724cd5a0481bbf9703b6c2266ae0 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":2},{"id":"7e39251ceb15fc3f481e998b00c52f10b4b0f003","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-08-10 08:14:15.000000000","message":"Patch Set 2: Code-Review-2","accounts_in_message":[],"_revision_number":2},{"id":"b6fe0aa94367b17906d7e711039ee4415280b90c","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-08-10 08:15:36.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"56113c49fb4aee1f0623f14992f674890af961a0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36229,"name":"Natz cheng","display_name":"Natzcheng","email":"chenggang01@inspur.com","username":"Natzcheng"},"date":"2023-08-10 08:23:08.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bfe0569de640bbac4a34aae83cb03ef8bfb8dbb3","revisions":{"e3f5604090709c38ddfc541e2dbde8fffb9770a1":{"kind":"REWORK","_number":1,"created":"2023-08-10 07:01:27.000000000","uploader":{"_account_id":36229,"name":"Natz cheng","display_name":"Natzcheng","email":"chenggang01@inspur.com","username":"Natzcheng"},"ref":"refs/changes/15/891015/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/15/891015/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/15/891015/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/15/891015/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/15/891015/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/15/891015/1"}}},"commit":{"parents":[{"commit":"d3d07cfd5aa574f1abeb5bbac9e908ca8dc208d4","subject":"update anomaly rule list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/d3d07cfd5aa574f1abeb5bbac9e908ca8dc208d4"}]}],"author":{"name":"Natzcheng","email":"chenggang01@inspur.com","date":"2023-08-10 07:01:02.000000000","tz":480},"committer":{"name":"Natzcheng","email":"chenggang01@inspur.com","date":"2023-08-10 07:01:02.000000000","tz":480},"subject":"support request parameter log_type","message":"support request parameter log_type\n\nChange-Id: I46f0637f41bec279166c1fe9eb34ae420e3e715c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/e3f5604090709c38ddfc541e2dbde8fffb9770a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/e3f5604090709c38ddfc541e2dbde8fffb9770a1"}]},"branch":"refs/heads/master"},"bfe0569de640bbac4a34aae83cb03ef8bfb8dbb3":{"kind":"REWORK","_number":2,"created":"2023-08-10 07:02:26.000000000","uploader":{"_account_id":36229,"name":"Natz cheng","display_name":"Natzcheng","email":"chenggang01@inspur.com","username":"Natzcheng"},"ref":"refs/changes/15/891015/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/15/891015/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/15/891015/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/15/891015/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/15/891015/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/15/891015/2"}}},"commit":{"parents":[{"commit":"d3d07cfd5aa574f1abeb5bbac9e908ca8dc208d4","subject":"update anomaly rule list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/d3d07cfd5aa574f1abeb5bbac9e908ca8dc208d4"}]}],"author":{"name":"Natzcheng","email":"chenggang01@inspur.com","date":"2023-08-10 07:01:02.000000000","tz":480},"committer":{"name":"Natz cheng","email":"chenggang01@inspur.com","date":"2023-08-10 07:02:21.000000000","tz":0},"subject":"support request parameter log_type","message":"support request parameter log_type\n\nChange-Id: I46f0637f41bec279166c1fe9eb34ae420e3e715c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/bfe0569de640bbac4a34aae83cb03ef8bfb8dbb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/bfe0569de640bbac4a34aae83cb03ef8bfb8dbb3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
