)]}'
{"id":"openstack%2Fvenus~892446","triplet_id":"openstack%2Fvenus~master~Ib8d6967f657fa90e71030efdf29f45f1a1879946","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"23303":{"account":{"_account_id":23303,"name":"howardlee","email":"lihongweibj@inspur.com","username":"howardlee"},"last_update":"2023-08-23 05:37:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib8d6967f657fa90e71030efdf29f45f1a1879946","subject":"fix contributing.rst text content display bug","status":"MERGED","created":"2023-08-23 02:57:15.000000000","updated":"2023-08-23 05:38:03.000000000","submitted":"2023-08-23 05:37:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892446","meta_rev_id":"fd8e852ce0e98da83a9dfa1557992065bef01647","_number":892446,"virtual_id_number":892446,"owner":{"_account_id":23303,"name":"howardlee","email":"lihongweibj@inspur.com","username":"howardlee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-23 05:37:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-23 05:33:06.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-23 05:33:06.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-23 03:08:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-23 05:33:06.000000000","updated_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"}],"messages":[{"id":"75b0f9e89dcf0220846ca49e4577e9de3a6a427f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23303,"name":"howardlee","email":"lihongweibj@inspur.com","username":"howardlee"},"date":"2023-08-23 02:57:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8035cff7ed5cabeadf261016211812a8949ed3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 03:08:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d32d0ebc54244099a62f3c0ae76fffa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d5f4a6d5a0447848397ac57be1da837 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89215f1599014af781730e308f177635 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":1},{"id":"7c477be5b56a447a347a66653ee50c2c445420f7","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-08-23 05:33:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"28b347240e26e2dfd6fc284fa19ae007541f77a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 05:33:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d89470ecad34df2ac1e34d1d79b35420d7b95aef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 05:37:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/848aeb2114964839a534c4a4eecebcf0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8f2e160c4314d8ebb16e0780679b647 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3346b2e551db408eb95db1f7512b8c68 : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":1},{"id":"e4738e435a9c92dd108a97f154b010838e7cef33","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 05:37:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fd8e852ce0e98da83a9dfa1557992065bef01647","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 05:38:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/581e82f361af4fbd944868e66ef82a3a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7582e92b0ade4dad9867eeb81071c98c : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dac967ed140df1255053bf2743a529166ed5a738","revisions":{"dac967ed140df1255053bf2743a529166ed5a738":{"kind":"REWORK","_number":1,"created":"2023-08-23 02:57:15.000000000","uploader":{"_account_id":23303,"name":"howardlee","email":"lihongweibj@inspur.com","username":"howardlee"},"ref":"refs/changes/46/892446/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/46/892446/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/46/892446/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/46/892446/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/46/892446/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/46/892446/1"}}},"commit":{"parents":[{"commit":"14928a697d760beab23610af39f93ff3cadc283c","subject":"Unify interface description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/14928a697d760beab23610af39f93ff3cadc283c"}]}],"author":{"name":"howardlee","email":"lihongweibj@inspur.com","date":"2023-08-23 02:52:38.000000000","tz":480},"committer":{"name":"howardlee","email":"lihongweibj@inspur.com","date":"2023-08-23 02:52:38.000000000","tz":480},"subject":"fix contributing.rst text content display bug","message":"fix contributing.rst text content display bug\n\nThe text content in the contributing.rst file display incorrectly,\nthis patch will fix it.\n\nChange-Id: Ib8d6967f657fa90e71030efdf29f45f1a1879946\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/dac967ed140df1255053bf2743a529166ed5a738"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/dac967ed140df1255053bf2743a529166ed5a738"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
