)]}'
{"id":"openstack%2Fvenus~893124","triplet_id":"openstack%2Fvenus~master~Iba364a19c408598f103f60ed27a0f47e8d8a7e84","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"23303":{"account":{"_account_id":23303,"name":"howardlee","email":"lihongweibj@inspur.com","username":"howardlee"},"last_update":"2023-08-30 05:50:49.000000000","reason":"Change was submitted"},"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2023-08-30 05:46:46.000000000","reason":"\u003cGERRIT_ACCOUNT_27565\u003e replied on the change","reason_account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}},"hashtags":[],"change_id":"Iba364a19c408598f103f60ed27a0f47e8d8a7e84","subject":"Using latest sphinx release version","status":"MERGED","created":"2023-08-30 02:59:52.000000000","updated":"2023-08-30 05:51:51.000000000","submitted":"2023-08-30 05:50:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893124","meta_rev_id":"6861669af6e8d7f5e2a5bcbaa4407c9327f787ce","_number":893124,"virtual_id_number":893124,"owner":{"_account_id":23303,"name":"howardlee","email":"lihongweibj@inspur.com","username":"howardlee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-30 05:50:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-30 05:46:46.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-30 05:46:46.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-30 03:18:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-30 03:26:06.000000000","updated_by":{"_account_id":23303,"name":"howardlee","email":"lihongweibj@inspur.com","username":"howardlee"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"}],"messages":[{"id":"06cfd7eddfcf81a022464d72a1cee6c6e28d504c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23303,"name":"howardlee","email":"lihongweibj@inspur.com","username":"howardlee"},"date":"2023-08-30 02:59:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8bb2e5855b246e0ffdd18640cd4748de4374686f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23303,"name":"howardlee","email":"lihongweibj@inspur.com","username":"howardlee"},"date":"2023-08-30 03:03:07.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ed6bb6738c7d314d3f4374b1fe5bfd3d4f2ee18c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23303,"name":"howardlee","email":"lihongweibj@inspur.com","username":"howardlee"},"date":"2023-08-30 03:06:21.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"2ef18c81ad6a7b9a30f0ea53fb14ea2a6b4d14de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 03:18:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1587537f233144c0ad492b1f36ec2aea\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2d9d232823bf4ae696f3d26bd113b386 : SUCCESS in 3m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da5e6905aad7477a9d859d8957f1f861 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/084c24aea77440a5ba666b4441e3f5e0 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f52cb116b66420085471c856a5c755a : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":3},{"id":"2a3e3b3b0ec66c2d173de294419e6649d53ab8da","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-08-30 05:46:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"27dab43e90ebf89818494355fd5d89bcb019d0d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 05:47:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4af8724f81f8503e672c783f8b0d7e592658c51e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 05:50:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/155c0498a07247b08779cf2c213aa07f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/127f9f9f552145239061cd8ae152265e : SUCCESS in 2m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fab3e58bbeda49158ac57fcdcb5a3ebf : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6672a09038c4bf2829e1c16ba29d7c7 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59da99cba6d1477699b91786943caa82 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":3},{"id":"6bea67b6688f41c060d401ada7b58709b556e982","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 05:50:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6861669af6e8d7f5e2a5bcbaa4407c9327f787ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 05:51:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09af64e1dc94438392b29d2889f1022a\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/609fd517fb5449e396c86857b4fbe2c8 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a04e4509e1f04307bc5db313b3ae8216 : SUCCESS in 41s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be694b3c8a744f5983338a08ac63261e : SUCCESS in 33s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f24112d1cd9b761fee13f28361a94ec0245844f8","revisions":{"aac6257a87ffc3a251d76c49184add659e84f586":{"kind":"REWORK","_number":1,"created":"2023-08-30 02:59:52.000000000","uploader":{"_account_id":23303,"name":"howardlee","email":"lihongweibj@inspur.com","username":"howardlee"},"ref":"refs/changes/24/893124/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/24/893124/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/24/893124/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/24/893124/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/24/893124/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/24/893124/1"}}},"commit":{"parents":[{"commit":"dac967ed140df1255053bf2743a529166ed5a738","subject":"fix contributing.rst text content display bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/dac967ed140df1255053bf2743a529166ed5a738"}]}],"author":{"name":"howardlee","email":"lihongweibj@inspur.com","date":"2023-08-30 02:53:49.000000000","tz":480},"committer":{"name":"howardlee","email":"lihongweibj@inspur.com","date":"2023-08-30 02:53:49.000000000","tz":480},"subject":"Using latest sphinx version","message":"Using latest sphinx version\n\nopenstackdocstheme has released version 3.2.0 [1], change sphinx to use the\nlatest version.\n\n[1] https://opendev.org/openstack/openstackdocstheme/releases/tag/3.2.0\n\nChange-Id: Iba364a19c408598f103f60ed27a0f47e8d8a7e84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/aac6257a87ffc3a251d76c49184add659e84f586"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/aac6257a87ffc3a251d76c49184add659e84f586"}]},"branch":"refs/heads/master"},"dbc1380e9f248f2139d02cc7a881bab58db4120a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-08-30 03:03:07.000000000","uploader":{"_account_id":23303,"name":"howardlee","email":"lihongweibj@inspur.com","username":"howardlee"},"ref":"refs/changes/24/893124/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/24/893124/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/24/893124/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/24/893124/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/24/893124/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/24/893124/2"}}},"commit":{"parents":[{"commit":"dac967ed140df1255053bf2743a529166ed5a738","subject":"fix contributing.rst text content display bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/dac967ed140df1255053bf2743a529166ed5a738"}]}],"author":{"name":"howardlee","email":"lihongweibj@inspur.com","date":"2023-08-30 02:53:49.000000000","tz":480},"committer":{"name":"howardlee","email":"lihongweibj@inspur.com","date":"2023-08-30 03:03:06.000000000","tz":0},"subject":"Using latest sphinx release version","message":"Using latest sphinx release version\n\nopenstackdocstheme has released version 3.2.0 [1], change sphinx to use the\nlatest version.\n\n[1] https://opendev.org/openstack/openstackdocstheme/releases/tag/3.2.0\n\nChange-Id: Iba364a19c408598f103f60ed27a0f47e8d8a7e84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/dbc1380e9f248f2139d02cc7a881bab58db4120a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/dbc1380e9f248f2139d02cc7a881bab58db4120a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f24112d1cd9b761fee13f28361a94ec0245844f8":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-08-30 03:06:21.000000000","uploader":{"_account_id":23303,"name":"howardlee","email":"lihongweibj@inspur.com","username":"howardlee"},"ref":"refs/changes/24/893124/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/24/893124/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/24/893124/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/24/893124/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/24/893124/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/24/893124/3"}}},"commit":{"parents":[{"commit":"dac967ed140df1255053bf2743a529166ed5a738","subject":"fix contributing.rst text content display bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/dac967ed140df1255053bf2743a529166ed5a738"}]}],"author":{"name":"howardlee","email":"lihongweibj@inspur.com","date":"2023-08-30 02:53:49.000000000","tz":480},"committer":{"name":"howardlee","email":"lihongweibj@inspur.com","date":"2023-08-30 03:06:20.000000000","tz":0},"subject":"Using latest sphinx release version","message":"Using latest sphinx release version\n\nopenstackdocstheme has released version 3.2.0 [1], change sphinx to use the latest version.\n\n[1] https://opendev.org/openstack/openstackdocstheme/releases/tag/3.2.0\n\nChange-Id: Iba364a19c408598f103f60ed27a0f47e8d8a7e84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/f24112d1cd9b761fee13f28361a94ec0245844f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/f24112d1cd9b761fee13f28361a94ec0245844f8"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
