)]}'
{"id":"openstack%2Fvenus~893327","triplet_id":"openstack%2Fvenus~master~I35adb4eca4a6c710fdff75d7dc3c0685882e7c09","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2023-09-05 01:58:33.000000000","reason":"\u003cGERRIT_ACCOUNT_27565\u003e replied on the change","reason_account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},"33950":{"account":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"last_update":"2023-09-05 02:02:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I35adb4eca4a6c710fdff75d7dc3c0685882e7c09","subject":"Add response sample of API \u0027add anomaly rule\u0027","status":"MERGED","created":"2023-09-05 01:31:12.000000000","updated":"2023-09-05 02:13:42.000000000","submitted":"2023-09-05 02:02:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893327","meta_rev_id":"c3e1723c0c8fe8a583bc9fd100560578929c35f5","_number":893327,"virtual_id_number":893327,"owner":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-05 02:02:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-05 01:58:33.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-05 01:58:33.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-05 01:36:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-05 01:38:21.000000000","updated_by":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"}],"messages":[{"id":"80cd96e0ecd839cd1d04f1277ecaa7189de1d12f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"date":"2023-09-05 01:31:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1493d3b5313af484b5b5e3b86f2068661674f69d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 01:36:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c6e47f22868466fb83a24ef899eaa9e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/18f6c18a9a3c41159fd05f11d0347c3f : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00718e41094a4f82bb426842b823c1fa : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5571a854152d45cf968d103a543533f7 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":1},{"id":"2b1ce1f87abe9c2bad73b942e6857514a8a28162","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-09-05 01:58:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aca7ebfaac3706aa60f20ec840feca4d9f758bfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 01:58:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"308c7731986d04e684b24b17ce6e138a8ba232be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 02:02:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d43988f37174184befd687fa2cde13d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a8ce20b44ab24f9e8f501786bba31a92 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/656a70302089451097bc41a6271aa27a : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ed856c7fe6742a1baaf3f2a29db4a78 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":1},{"id":"500110ce53af39644df30594ded9b8bbfecd0830","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 02:02:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c3e1723c0c8fe8a583bc9fd100560578929c35f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 02:13:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2d5e99a74a14c3f88511f5678d4509a\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/705e4c9859e849baa768f860f283e65b : SUCCESS in 56s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bc85ea3353b44e4b5057f6b6f8590c6 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4549137b8f90d77e21d8835b61f68ef52b5d5aee","revisions":{"4549137b8f90d77e21d8835b61f68ef52b5d5aee":{"kind":"REWORK","_number":1,"created":"2023-09-05 01:31:12.000000000","uploader":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"ref":"refs/changes/27/893327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/27/893327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/27/893327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/27/893327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/27/893327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/27/893327/1"}}},"commit":{"parents":[{"commit":"664f424525935e0e750bd940375f83c090300c2e","subject":"Merge \"add api doc: add anomaly rule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/664f424525935e0e750bd940375f83c090300c2e"}]}],"author":{"name":"maliangyi","email":"maliangyi@inspur.com","date":"2023-09-05 01:28:22.000000000","tz":480},"committer":{"name":"maliangyi","email":"maliangyi@inspur.com","date":"2023-09-05 01:30:57.000000000","tz":480},"subject":"Add response sample of API \u0027add anomaly rule\u0027","message":"Add response sample of API \u0027add anomaly rule\u0027\n\nChange-Id: I35adb4eca4a6c710fdff75d7dc3c0685882e7c09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/4549137b8f90d77e21d8835b61f68ef52b5d5aee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/4549137b8f90d77e21d8835b61f68ef52b5d5aee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
