)]}'
{"id":"openstack%2Fvenus~894881","triplet_id":"openstack%2Fvenus~master~I16e2773a5570d3000c0ac57c3a4d0df7399978fd","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2023-09-14 06:46:47.000000000","reason":"\u003cGERRIT_ACCOUNT_27565\u003e replied on the change","reason_account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},"33950":{"account":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"last_update":"2023-09-14 06:56:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I16e2773a5570d3000c0ac57c3a4d0df7399978fd","subject":"Add request sample of API \u0027delete anomaly record\u0027","status":"MERGED","created":"2023-09-13 06:36:58.000000000","updated":"2023-09-14 06:56:31.000000000","submitted":"2023-09-14 06:56:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894881","meta_rev_id":"ef48bed14f4f23bf734564ff8dbcfaeabe15bddf","_number":894881,"virtual_id_number":894881,"owner":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-14 06:56:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-14 06:46:47.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-14 06:46:47.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-13 06:45:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-13 06:54:39.000000000","updated_by":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"}],"messages":[{"id":"c73acd910353a871abb21be750da7ed662730dd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"date":"2023-09-13 06:36:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f8f7e696e224e145ee90987179e7bad7ca7a953","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 06:45:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e65a0b335bc4393bbb9f1773a05a8a2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/79dcf4bf40e443f5a9995621042727a4 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5c1a71b666d49e0901332e7ad30442b : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38660cafb1a94a15b53a2c7bb66aacfd : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"088a77a613611f1ddf2785f9238d2dbf33600d73","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-09-14 06:46:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5cb542b6508fe58d762e9d3c164684f88dac7831","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 06:47:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e732a7e8250ff19e522de3c5b3bcebd73780c027","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 06:56:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4a4e13584ab4609b1b05ebfa35a0839\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/748593a8c935478f9a26d19acb1c55b9 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1af17b019d5b40e0b0e68b9befd61512 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e08a6bc644b4ee3be6c07938546a50d : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"ef48bed14f4f23bf734564ff8dbcfaeabe15bddf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 06:56:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"abffa881a88e141f332f94dde53bf1b514f4b24a","revisions":{"abffa881a88e141f332f94dde53bf1b514f4b24a":{"kind":"REWORK","_number":1,"created":"2023-09-13 06:36:58.000000000","uploader":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"ref":"refs/changes/81/894881/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/81/894881/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/81/894881/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/81/894881/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/81/894881/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/81/894881/1"}}},"commit":{"parents":[{"commit":"d34bbb207f6678f3bced3f1b4fe8acc2516ee9fd","subject":"Adding uuid param for callchain api interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/d34bbb207f6678f3bced3f1b4fe8acc2516ee9fd"}]}],"author":{"name":"maliangyi","email":"maliangyi@inspur.com","date":"2023-09-13 06:34:20.000000000","tz":480},"committer":{"name":"maliangyi","email":"maliangyi@inspur.com","date":"2023-09-13 06:36:45.000000000","tz":480},"subject":"Add request sample of API \u0027delete anomaly record\u0027","message":"Add request sample of API \u0027delete anomaly record\u0027\n\nChange-Id: I16e2773a5570d3000c0ac57c3a4d0df7399978fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/abffa881a88e141f332f94dde53bf1b514f4b24a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/abffa881a88e141f332f94dde53bf1b514f4b24a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
