)]}'
{"id":"openstack%2Fvenus~894883","triplet_id":"openstack%2Fvenus~master~Ic910045e5209791193d8b76c2a395684bbdeff50","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"33950":{"account":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"last_update":"2023-09-14 07:58:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic910045e5209791193d8b76c2a395684bbdeff50","subject":"update samples","status":"MERGED","created":"2023-09-14 07:40:43.000000000","updated":"2023-09-14 07:59:11.000000000","submitted":"2023-09-14 07:58:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894883","meta_rev_id":"53fb812b9cb698b465eae908dcbc0965929cb988","_number":894883,"virtual_id_number":894883,"owner":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-14 07:58:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-14 07:52:12.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-14 07:52:12.000000000","_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-14 07:49:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-14 07:52:12.000000000","updated_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"reviewer":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"state":"REVIEWER"}],"messages":[{"id":"8e35960fe1da6e0f3d4804d84ce3ae9acf93a363","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"date":"2023-09-14 07:40:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8b81570d1bcd14f3c2252739bbe20307b9a721d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 07:49:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/046a37f84afb4cf8a31ba9b459aca47a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/31b6c6a3965a4ffaac1480f52bc86a98 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc5f8a09b88745b7aeb4375eb8b7ca05 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09841a5bcd4c47c2bc2777423f2e8863 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4b705d9579e5ce2cac7be7a17b330449d89cdd86","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2023-09-14 07:52:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90df1a9a73c859f49d69b41fd08abcaf879193b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 07:52:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7506b9ab45138842b024474d16fcb6b2711c1fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 07:58:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66b5950aa7684f20aab31450ba5bd4d0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f6a04b66be224ffaaa5bde2bcf402b7e : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa1d9df3785d4bd393fc527073b33256 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90948368f1e34c65a076396a4c6b302e : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"6cb7db1f4007f4cbcfa5f313564f9a7afe31703e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 07:58:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"53fb812b9cb698b465eae908dcbc0965929cb988","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 07:59:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4482219b2b00411787ccc550a93769a7\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/f6f944eb72fc4f35b92b072e8ca938a3 : SUCCESS in 42s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a2604552ee246f09fc4b3d851848739 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"caaebe69fd5d01914411eee6f05de1be59fbcc76","revisions":{"caaebe69fd5d01914411eee6f05de1be59fbcc76":{"kind":"REWORK","_number":1,"created":"2023-09-14 07:40:43.000000000","uploader":{"_account_id":33950,"name":"Liangyi Ma","email":"maliangyi@inspur.com","username":"maliangyi"},"ref":"refs/changes/83/894883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/83/894883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/83/894883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/83/894883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/83/894883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/83/894883/1"}}},"commit":{"parents":[{"commit":"0feeee3b19dfb572a04456a080fdeb6bb77d4d1a","subject":"Merge \"Fix params to required of api \u0027add anomaly rule\u0027 in api-ref\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/0feeee3b19dfb572a04456a080fdeb6bb77d4d1a"}]}],"author":{"name":"maliangyi","email":"maliangyi@inspur.com","date":"2023-09-14 07:37:55.000000000","tz":480},"committer":{"name":"maliangyi","email":"maliangyi@inspur.com","date":"2023-09-14 07:37:55.000000000","tz":480},"subject":"update samples","message":"update samples\n\nChange-Id: Ic910045e5209791193d8b76c2a395684bbdeff50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/caaebe69fd5d01914411eee6f05de1be59fbcc76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/caaebe69fd5d01914411eee6f05de1be59fbcc76"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
