)]}'
{"id":"openstack%2Fvenus~899100","triplet_id":"openstack%2Fvenus~master~I48c24f54352d2ebdcb65f518a2ed5a30b5efe545","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"35819":{"account":{"_account_id":35819,"name":"zhuyawei","display_name":"zhuyawei","email":"zhuyawei@inspur.com","username":"zhuyawei"},"last_update":"2023-12-21 08:39:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I48c24f54352d2ebdcb65f518a2ed5a30b5efe545","subject":"Supplement a unit test case for \u0027validate_key_names\u0027","status":"MERGED","created":"2023-10-24 01:23:11.000000000","updated":"2023-12-21 08:40:18.000000000","submitted":"2023-12-21 08:39:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899100","meta_rev_id":"ce9f5723468464f90832514b0d7e7bb3f1c53ba6","_number":899100,"virtual_id_number":899100,"owner":{"_account_id":35819,"name":"zhuyawei","display_name":"zhuyawei","email":"zhuyawei@inspur.com","username":"zhuyawei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-21 08:39:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2023-12-21 08:05:00.000000000","_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":1,"date":"2023-12-21 08:05:00.000000000","_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-24 01:33:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-21 08:05:00.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"}],"messages":[{"id":"a200186d809c52fd7cd913651e778f49e74907d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35819,"name":"zhuyawei","display_name":"zhuyawei","email":"zhuyawei@inspur.com","username":"zhuyawei"},"date":"2023-10-24 01:23:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fcc0f4acf7867201e25655ad9906082d5c6ebbb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 01:33:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb2c6808e0be4978b7fed5694cc3204b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fae025e7b933495bb8f2a3f49dc19f70 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9bae11b00584e03ba1059b8a41d1b4c : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b21f1130cf6d44dfb89a48b628821489 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3489f167403b409ea3a8f94cf964e56a : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":1},{"id":"91827447d96ebab730c862dcf496d5884da59515","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2023-12-21 08:05:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7e8887df12ec024a3fe7af40dd3930917acfe614","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 08:05:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dcb1ed11e4ca0f61dcfacfa3219a565af83ba9e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 08:39:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f86ca5764142433c8bbf537185669273\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e627920ae41d4b6aba6ee2b6a3d25cc7 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe7322b5da1c485ebfbce97de54b2306 : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab4718e92ce24824983630a97e7bccf9 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30cc18783615471cbb3b07f073213025 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":1},{"id":"14512fccd6ef91ff1562bab06bda014fc7528889","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 08:39:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ce9f5723468464f90832514b0d7e7bb3f1c53ba6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 08:40:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb6e3d7c4d2b4873b872a275487bcd88\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cefe034f79904598b97d654f40c98704 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"66876bdcf10c36e782831e270436ebdc34e517b9","revisions":{"66876bdcf10c36e782831e270436ebdc34e517b9":{"kind":"REWORK","_number":1,"created":"2023-10-24 01:23:11.000000000","uploader":{"_account_id":35819,"name":"zhuyawei","display_name":"zhuyawei","email":"zhuyawei@inspur.com","username":"zhuyawei"},"ref":"refs/changes/00/899100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/00/899100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/00/899100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/00/899100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/00/899100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/00/899100/1"}}},"commit":{"parents":[{"commit":"b590c1a84f475aeccf6c63d089c7e40e99fd5994","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/b590c1a84f475aeccf6c63d089c7e40e99fd5994"}]}],"author":{"name":"zhuyawei","email":"zhuyawei@inspur.com","date":"2023-10-24 01:02:21.000000000","tz":480},"committer":{"name":"zhuyawei","email":"zhuyawei@inspur.com","date":"2023-10-24 01:02:21.000000000","tz":480},"subject":"Supplement a unit test case for \u0027validate_key_names\u0027","message":"Supplement a unit test case for \u0027validate_key_names\u0027\n\nthe test case for api.common\n\nChange-Id: I48c24f54352d2ebdcb65f518a2ed5a30b5efe545\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/66876bdcf10c36e782831e270436ebdc34e517b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/66876bdcf10c36e782831e270436ebdc34e517b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
