)]}'
{"id":"openstack%2Fvenus~904191","triplet_id":"openstack%2Fvenus~master~I0aebcaf96f40dbdd523c0c2b78095d50f4d9e76b","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"32148":{"account":{"_account_id":32148,"name":"liuyuanfeng","email":"liuyuanfeng@inspur.com","username":"liuyuanfeng"},"last_update":"2024-01-03 07:12:00.000000000","reason":"Change was submitted"},"30455":{"account":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"last_update":"2024-01-03 06:24:34.000000000","reason":"\u003cGERRIT_ACCOUNT_30455\u003e replied on the change","reason_account":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}},"hashtags":[],"change_id":"I0aebcaf96f40dbdd523c0c2b78095d50f4d9e76b","subject":"add WSGI server start log","status":"MERGED","created":"2023-12-29 02:19:39.000000000","updated":"2024-01-03 07:13:34.000000000","submitted":"2024-01-03 07:12:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904191","meta_rev_id":"b98bad1a0ffadd597587550641a34b9012d5b12d","_number":904191,"virtual_id_number":904191,"owner":{"_account_id":32148,"name":"liuyuanfeng","email":"liuyuanfeng@inspur.com","username":"liuyuanfeng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-03 07:12:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":2,"date":"2024-01-03 06:24:34.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":1,"date":"2024-01-03 06:24:34.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-29 03:01:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-03 06:18:57.000000000","updated_by":{"_account_id":32148,"name":"liuyuanfeng","email":"liuyuanfeng@inspur.com","username":"liuyuanfeng"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"}],"messages":[{"id":"aaa867bd6714ab528fb6ad18c9e135e6a448eeea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32148,"name":"liuyuanfeng","email":"liuyuanfeng@inspur.com","username":"liuyuanfeng"},"date":"2023-12-29 02:19:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee40f278e554c556b9f81020e1be3dd4838099fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-29 03:01:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0c5813325064ff5b85b00d5ebfea4b5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e6cd15749a644569cca6c468659f4ed : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5daaec31b8e04fcab22b38fa582cf239 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ac925b9e71443298b78e167e2ca4c21 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/971259981c3a45f5b39c993e223a423e : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6da2883d22d17ebfe9e4dcd61e8d7487696b43cc","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2024-01-03 06:24:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"091a027929bc9a0669f71edf700089615e52e99a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 07:02:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"99b70c11dd77870d5419878be62313eb2cea07ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 07:12:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21380f99309246baa96174e02b9ca8e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8e62dcb060844ca9ce99ab51ab5ced7 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/508046bc29db4d7e8cd5a69ed41fd0a6 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/760bcb23daf24733b422680e210f40e5 : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/54058ec477be4d08a81ea1a1cf713bfe : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"ca48e59cd22c3469feebf24611609a8f1223e5a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 07:12:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b98bad1a0ffadd597587550641a34b9012d5b12d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 07:13:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00ff4e3b99944a35ab6e0981de5840a2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8227362f04b64b4bbf7bcc168a5abf42 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da298d7835197c8648aa972e48e9e14669d75f34","revisions":{"da298d7835197c8648aa972e48e9e14669d75f34":{"kind":"REWORK","_number":1,"created":"2023-12-29 02:19:39.000000000","uploader":{"_account_id":32148,"name":"liuyuanfeng","email":"liuyuanfeng@inspur.com","username":"liuyuanfeng"},"ref":"refs/changes/91/904191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/91/904191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/91/904191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/91/904191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/91/904191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/91/904191/1"}}},"commit":{"parents":[{"commit":"1105c801e19360134891fa14db7787889880d3fa","subject":"add some Log in manager.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/1105c801e19360134891fa14db7787889880d3fa"}]}],"author":{"name":"liuyuanfeng","email":"liuyuanfeng@inspur.com","date":"2023-12-29 02:21:35.000000000","tz":480},"committer":{"name":"liuyuanfeng","email":"liuyuanfeng@inspur.com","date":"2023-12-29 02:21:35.000000000","tz":480},"subject":"add WSGI server start log","message":"add WSGI server start log\n\nChange-Id: I0aebcaf96f40dbdd523c0c2b78095d50f4d9e76b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/da298d7835197c8648aa972e48e9e14669d75f34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/da298d7835197c8648aa972e48e9e14669d75f34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
