)]}'
{"id":"openstack%2Fvenus~904980","triplet_id":"openstack%2Fvenus~master~If9b3d712cfb5e50b28c5e74cbfa0e2d5272d3d81","project":"openstack/venus","branch":"master","attention_set":{},"removed_from_attention_set":{"27565":{"account":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"last_update":"2024-01-17 01:00:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If9b3d712cfb5e50b28c5e74cbfa0e2d5272d3d81","subject":"delete es index over maxsize","status":"MERGED","created":"2024-01-16 10:59:05.000000000","updated":"2024-01-17 01:01:21.000000000","submitted":"2024-01-17 01:00:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904980","meta_rev_id":"0bfc2ca228a7c7775515af738407efad4e7143e4","_number":904980,"virtual_id_number":904980,"owner":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-17 01:00:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":2,"date":"2024-01-17 00:49:35.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"all":[{"value":1,"date":"2024-01-17 00:49:35.000000000","_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-16 11:08:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-17 00:49:35.000000000","updated_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"reviewer":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"state":"REVIEWER"}],"messages":[{"id":"ac1e46d4589e306c0cb167beddcbbdf538fcc757","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2024-01-16 10:59:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f8f6140bc99559bfe0618313c65dfaa4f50dae2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 11:08:47.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2d93f60e28f449493e4c03eb1cd5f6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56d076b962124effa6953044d2bbacfd : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e4ba53ca0b3461889a8b44eb5ead602 : FAILURE in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c188ec93c3bf40b3838342a8d0ad2e7f : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50bb9efb0b384b32ae8cd87bd7c777a2 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"14076d29efa27a72aec9e4af253b6eda9ed1ad02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2024-01-16 11:13:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"19a76aa4272de010292cbef6687d284163c9a91d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 11:24:14.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c511c898abb54e70ae4c322fd5bb416b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6198e4ad8f64496598f43dba22ba7d54 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f82fd6ef7434429888bad579c899748 : FAILURE in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a766436c5ad46b88c10c28ed92f0053 : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d1edeeb3ef34ff784c8f44751d60a3b : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":2},{"id":"0077b2a80a34979033b1c849a3788f87b3f6d61f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"date":"2024-01-16 11:25:33.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ce6aee5eb1aab93e7d5a84f8daf97312c5194122","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 11:34:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b906f625e88d4586ae31386c2935f7c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01929107b62b449085202db3b6847c8f : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca2350e4458648f5a901fcf29ab63f40 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/851083814ec141558c184405e22a5023 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/35baf6495f604e14a9c3c8d7d636054f : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":3},{"id":"bf2b8f7aaecdc363e7bd405ea4850a3785d83bd3","author":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"},"date":"2024-01-17 00:49:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"047ff1c316416468f69662a6a6353fbd0b21d5d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 00:49:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"850bf0aaad77caaf3942bd1190e7c4325668e45c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 01:00:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff34381caf3f48adb848ddb9f93bdf19\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/612071186b5544229098b5f0ebef43a5 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1c7e8c4caa94f5eb9c4c0497c32040c : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa64e78a2b964b07bbf01aed902f05db : SUCCESS in 3m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5b9303c3c4174bd584370de18d3c270d : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":3},{"id":"6c19f24de709c017956914b78626faa5ec69494c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 01:00:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0bfc2ca228a7c7775515af738407efad4e7143e4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 01:01:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d5a8e7dcc594ba3ad703691c03da680\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/820f54933b884b93971d89e5c10a814d : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ea0d95bf7f27dad44733c97ba26e0ac1ce4e73bd","revisions":{"65874e042939560441b12bc769062f062e81f3cb":{"kind":"REWORK","_number":1,"created":"2024-01-16 10:59:05.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/80/904980/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/80/904980/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/80/904980/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/80/904980/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/80/904980/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/80/904980/1"}}},"commit":{"parents":[{"commit":"4d673fb1fd0c420ae42371a231bca7042f3293b1","subject":"add new version sql update file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/4d673fb1fd0c420ae42371a231bca7042f3293b1"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2024-01-16 10:58:50.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2024-01-16 10:58:50.000000000","tz":0},"subject":"delete es index over maxsize","message":"delete es index over maxsize\n\nChange-Id: If9b3d712cfb5e50b28c5e74cbfa0e2d5272d3d81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/65874e042939560441b12bc769062f062e81f3cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/65874e042939560441b12bc769062f062e81f3cb"}]},"branch":"refs/heads/master"},"2730e3b7b8921f2c6ade48ce3c8aea21d835792e":{"kind":"REWORK","_number":2,"created":"2024-01-16 11:13:15.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/80/904980/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/80/904980/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/80/904980/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/80/904980/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/80/904980/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/80/904980/2"}}},"commit":{"parents":[{"commit":"4d673fb1fd0c420ae42371a231bca7042f3293b1","subject":"add new version sql update file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/4d673fb1fd0c420ae42371a231bca7042f3293b1"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2024-01-16 10:58:50.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2024-01-16 11:12:57.000000000","tz":0},"subject":"delete es index over maxsize","message":"delete es index over maxsize\n\nChange-Id: If9b3d712cfb5e50b28c5e74cbfa0e2d5272d3d81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/2730e3b7b8921f2c6ade48ce3c8aea21d835792e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/2730e3b7b8921f2c6ade48ce3c8aea21d835792e"}]},"branch":"refs/heads/master"},"ea0d95bf7f27dad44733c97ba26e0ac1ce4e73bd":{"kind":"REWORK","_number":3,"created":"2024-01-16 11:25:33.000000000","uploader":{"_account_id":27565,"name":"inspurericzhang","email":"zhanglf01@inspur.com","username":"inspurericzhang"},"ref":"refs/changes/80/904980/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/venus","ref":"refs/changes/80/904980/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/venus refs/changes/80/904980/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/venus refs/changes/80/904980/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/venus refs/changes/80/904980/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/venus refs/changes/80/904980/3"}}},"commit":{"parents":[{"commit":"4d673fb1fd0c420ae42371a231bca7042f3293b1","subject":"add new version sql update file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/4d673fb1fd0c420ae42371a231bca7042f3293b1"}]}],"author":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2024-01-16 10:58:50.000000000","tz":0},"committer":{"name":"inspurericzhang","email":"zhanglf01@inspur.com","date":"2024-01-16 11:25:20.000000000","tz":0},"subject":"delete es index over maxsize","message":"delete es index over maxsize\n\nChange-Id: If9b3d712cfb5e50b28c5e74cbfa0e2d5272d3d81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ea0d95bf7f27dad44733c97ba26e0ac1ce4e73bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/venus/commit/ea0d95bf7f27dad44733c97ba26e0ac1ce4e73bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30455,"name":"jiasirui","email":"jiasirui@inspur.com","username":"JiaSiRui"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
