)]}'
{"id":"openstack%2Fvirtualbmc~384755","triplet_id":"openstack%2Fvirtualbmc~master~Iae1682c51d267c47ac1a0ba270cfa866bed265a5","project":"openstack/virtualbmc","branch":"master","topic":"MagicMock","hashtags":[],"change_id":"Iae1682c51d267c47ac1a0ba270cfa866bed265a5","subject":"Replace \u0027MagicMock\u0027 with \u0027Mock\u0027","status":"ABANDONED","created":"2016-10-11 01:13:20.000000000","updated":"2016-10-11 02:07:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f981f5eca5b5a9ea8baaa06d224cebec29f12a5d","_number":384755,"virtual_id_number":384755,"owner":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":1,"date":"2016-10-11 01:57:35.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":-2,"date":"2016-10-11 02:04:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-10-11 01:57:35.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-10-11 02:04:49.000000000","updated_by":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"reviewer":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"state":"REVIEWER"}],"messages":[{"id":"f672f05c0d56657027a744ad8b768addc458c8cc","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2016-10-11 01:13:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f0c82061e098c37ca0ef1516b7961dd7dfbca84","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-11 01:57:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-virtualbmc-docs-ubuntu-xenial http://docs-draft.openstack.org/55/384755/1/check/gate-virtualbmc-docs-ubuntu-xenial/562b01a//doc/build/html/ : SUCCESS in 2m 42s\n- gate-virtualbmc-pep8-ubuntu-xenial http://logs.openstack.org/55/384755/1/check/gate-virtualbmc-pep8-ubuntu-xenial/9ef4db1/ : SUCCESS in 2m 47s\n- gate-virtualbmc-python27-ubuntu-xenial http://logs.openstack.org/55/384755/1/check/gate-virtualbmc-python27-ubuntu-xenial/c15787d/ : SUCCESS in 3m 03s\n- gate-tempest-dsvm-ironic-pxe_ipmitool-virtualbmc-src-nv http://logs.openstack.org/55/384755/1/check/gate-tempest-dsvm-ironic-pxe_ipmitool-virtualbmc-src-nv/bbcf402/ : SUCCESS in 44m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"894f43d5a8979c144fb8bd465c7b3408cf35871e","author":{"_account_id":14760,"name":"John L. Villalovos","email":"openstack.org@sodarock.com","username":"jlvillal"},"date":"2016-10-11 02:04:49.000000000","message":"Patch Set 1: Code-Review-2\n\nPlease see:\n\nhttp://lists.openstack.org/pipermail/openstack-dev/2016-September/104178.html","accounts_in_message":[],"_revision_number":1},{"id":"9530912b4f5de58541e20bd747e5ca599b356296","author":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"date":"2016-10-11 02:07:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d776b20caf5d125ff85390bdd4d932d646e032b6","revisions":{"d776b20caf5d125ff85390bdd4d932d646e032b6":{"kind":"REWORK","_number":1,"created":"2016-10-11 01:13:20.000000000","uploader":{"_account_id":21387,"name":"gecong","email":"ge.cong@zte.com.cn","username":"gecong"},"ref":"refs/changes/55/384755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/virtualbmc","ref":"refs/changes/55/384755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/virtualbmc refs/changes/55/384755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/virtualbmc refs/changes/55/384755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/virtualbmc refs/changes/55/384755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/virtualbmc refs/changes/55/384755/1"}}},"commit":{"parents":[{"commit":"f3c05f9c011e947cac262ce27e0c9027eaa81698","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/virtualbmc/commit/f3c05f9c011e947cac262ce27e0c9027eaa81698"}]}],"author":{"name":"gecong1973","email":"ge.cong@zte.com.cn","date":"2016-10-11 01:26:05.000000000","tz":480},"committer":{"name":"gecong1973","email":"ge.cong@zte.com.cn","date":"2016-10-11 01:26:05.000000000","tz":480},"subject":"Replace \u0027MagicMock\u0027 with \u0027Mock\u0027","message":"Replace \u0027MagicMock\u0027 with \u0027Mock\u0027\n\nMagicMock magically allow things to pass. Any usage should\nbe replaced with the Mock class and explicit attributes\nshould be set on it. [1]\n[1] http://stackoverflow.com/questions/17181687/mock-vs-magicmock\n\nChange-Id: Iae1682c51d267c47ac1a0ba270cfa866bed265a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/virtualbmc/commit/d776b20caf5d125ff85390bdd4d932d646e032b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/virtualbmc/commit/d776b20caf5d125ff85390bdd4d932d646e032b6"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f3c05f9c011e947cac262ce27e0c9027eaa81698","is_merged_in_target_branch":true,"change_id":"Ia0415718a0a6e8a7e486714be95b1033674f4af1","change_number":373687,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
