)]}'
{"id":"openstack%2Fvitrage-dashboard~654465","triplet_id":"openstack%2Fvitrage-dashboard~master~I17138c5176ed0b0d767af7f5263105d389e1e241","project":"openstack/vitrage-dashboard","branch":"master","hashtags":[],"change_id":"I17138c5176ed0b0d767af7f5263105d389e1e241","subject":"Changes for url transition to OpenDev","status":"MERGED","created":"2019-04-22 14:30:14.000000000","updated":"2019-04-23 08:54:26.000000000","submitted":"2019-04-23 08:54:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654465-1556009666654-eaa21e4c","meta_rev_id":"af231f39f034fc9f7e65a42b0ff2d5c3bda99546","_number":654465,"virtual_id_number":654465,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2019-04-23 08:54:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"all":[{"value":2,"date":"2019-04-23 05:30:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":2,"date":"2019-04-23 08:43:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":1,"date":"2019-04-23 08:43:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 05:30:28.000000000","updated_by":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"reviewer":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"state":"REVIEWER"},{"updated":"2019-04-23 08:43:08.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-04-23 08:54:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"243ab678beffa2d1d14157070ca81b0ce99c3455","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-04-22 14:30:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c892010827d19d455a2500ffe18bbb1fbee4c81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 15:16:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/65/654465/1/check/test-release-openstack/f6f51f7/ : SUCCESS in 2m 17s\n- nodejs10-npm-run-lint http://logs.openstack.org/65/654465/1/check/nodejs10-npm-run-lint/2aadfc0/ : SUCCESS in 5m 39s (non-voting)\n- nodejs10-npm-run-test http://logs.openstack.org/65/654465/1/check/nodejs10-npm-run-test/2d04cdb/ : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/65/654465/1/check/openstack-tox-pep8/659b9f8/ : SUCCESS in 5m 34s\n- openstack-tox-py27 http://logs.openstack.org/65/654465/1/check/openstack-tox-py27/d028990/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/65/654465/1/check/openstack-tox-py36/6ea82e8/ : SUCCESS in 4m 27s\n- openstack-tox-py37 http://logs.openstack.org/65/654465/1/check/openstack-tox-py37/738a719/ : SUCCESS in 4m 27s\n- openstack-tox-docs http://logs.openstack.org/65/654465/1/check/openstack-tox-docs/96ce755/html/ : SUCCESS in 3m 22s\n- build-openstack-releasenotes http://logs.openstack.org/65/654465/1/check/build-openstack-releasenotes/f50501f/html/ : SUCCESS in 2m 43s\n- vitrage-dashboard-integration-tests http://logs.openstack.org/65/654465/1/check/vitrage-dashboard-integration-tests/2420eb4/ : SUCCESS in 44m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae049d21741dda1749f2741773aabc3c680850dd","author":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"date":"2019-04-23 05:30:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f1c1a60cd99edf88d6ef45cd8abe967c1a0a0219","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-04-23 08:43:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f278c70d5821e05cc29cb9e71c5656aea4c5dab3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 08:43:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d135e26329b4972ef177648aa71823e10913718","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 08:54:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"af231f39f034fc9f7e65a42b0ff2d5c3bda99546","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 08:54:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/65/654465/1/gate/test-release-openstack/29b9ce3/ : SUCCESS in 1m 43s\n- openstack-tox-pep8 http://logs.openstack.org/65/654465/1/gate/openstack-tox-pep8/b5e7f3a/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/65/654465/1/gate/openstack-tox-py27/1ff3321/ : SUCCESS in 4m 08s\n- openstack-tox-py36 http://logs.openstack.org/65/654465/1/gate/openstack-tox-py36/2b49a13/ : SUCCESS in 4m 10s\n- openstack-tox-py37 http://logs.openstack.org/65/654465/1/gate/openstack-tox-py37/a400d40/ : SUCCESS in 4m 44s\n- openstack-tox-docs http://logs.openstack.org/65/654465/1/gate/openstack-tox-docs/ef3aeab/html/ : SUCCESS in 3m 56s\n- build-openstack-releasenotes http://logs.openstack.org/65/654465/1/gate/build-openstack-releasenotes/bc8a980/html/ : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4ca4358c6e5866e2b3dfc83b901d13f456eac272","revisions":{"4ca4358c6e5866e2b3dfc83b901d13f456eac272":{"kind":"REWORK","_number":1,"created":"2019-04-22 14:30:14.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/65/654465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/65/654465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/65/654465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/65/654465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/65/654465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/65/654465/1"}}},"commit":{"parents":[{"commit":"4e84ad65d1e754c57a633932231677ceef4562d5","subject":"Drop nodejs4 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/4e84ad65d1e754c57a633932231677ceef4562d5"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2019-04-22 14:29:26.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2019-04-22 14:30:01.000000000","tz":0},"subject":"Changes for url transition to OpenDev","message":"Changes for url transition to OpenDev\n\nThis commit changes openstack.org URL to opendev.org URL\n\nChange-Id: I17138c5176ed0b0d767af7f5263105d389e1e241\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/4ca4358c6e5866e2b3dfc83b901d13f456eac272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/4ca4358c6e5866e2b3dfc83b901d13f456eac272"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
