)]}'
{"id":"openstack%2Fvitrage-dashboard~655151","triplet_id":"openstack%2Fvitrage-dashboard~master~If2152bd721e81cd5fdfa25af1b94e820f56c9f84","project":"openstack/vitrage-dashboard","branch":"master","topic":"rca","hashtags":[],"change_id":"If2152bd721e81cd5fdfa25af1b94e820f56c9f84","subject":"fix alarm rca view","status":"MERGED","created":"2019-04-23 13:31:45.000000000","updated":"2020-04-24 12:13:29.000000000","submitted":"2020-04-24 12:12:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655151-1587730337522-c713193e","meta_rev_id":"ee55a1a02b0f1f8b5da5b4e44376b3358fc0bb12","_number":655151,"virtual_id_number":655151,"owner":{"_account_id":29793,"name":"Noor Yameen","email":"noor.yameen@exalt.ps","username":"NoorYameen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-24 12:12:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":2,"date":"2020-04-24 12:04:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-24 11:15:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":2,"date":"2020-04-24 11:15:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":1,"date":"2020-04-24 12:04:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 12:01:52.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"},{"updated":"2019-05-23 13:34:40.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-07-18 06:59:20.000000000","updated_by":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"reviewer":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"state":"REVIEWER"},{"updated":"2020-04-24 12:04:46.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-04-24 12:12:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0551172a728f89c167e9d21ace590913333943d8","author":{"_account_id":29793,"name":"Noor Yameen","email":"noor.yameen@exalt.ps","username":"NoorYameen"},"date":"2019-04-23 13:31:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0cb87d7d44a47f46ea8be9f5b12bf532e40d33ee","author":{"_account_id":29793,"name":"Noor Yameen","email":"noor.yameen@exalt.ps","username":"NoorYameen"},"date":"2019-04-23 14:09:20.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"32d2a4c1af90dd805cff77a913a3edeffec8d98f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 15:38:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/51/655151/2/check/nodejs10-npm-run-lint/83b1bc6/ : SUCCESS in 4m 36s (non-voting)\n- nodejs10-npm-run-test http://logs.openstack.org/51/655151/2/check/nodejs10-npm-run-test/09e1205/ : FAILURE in 5m 01s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/51/655151/2/check/openstack-tox-pep8/1d29c32/ : SUCCESS in 4m 26s\n- openstack-tox-py27 http://logs.openstack.org/51/655151/2/check/openstack-tox-py27/85ce05f/ : SUCCESS in 3m 54s\n- openstack-tox-py36 http://logs.openstack.org/51/655151/2/check/openstack-tox-py36/386d4a7/ : SUCCESS in 4m 04s\n- openstack-tox-py37 http://logs.openstack.org/51/655151/2/check/openstack-tox-py37/b6dd6e0/ : SUCCESS in 4m 44s\n- openstack-tox-docs http://logs.openstack.org/51/655151/2/check/openstack-tox-docs/e668c8e/html/ : SUCCESS in 3m 10s\n- vitrage-dashboard-integration-tests http://logs.openstack.org/51/655151/2/check/vitrage-dashboard-integration-tests/4a07e56/ : SUCCESS in 42m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a8058e1e6364b82be5680bcb25020e1ae5687556","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2019-04-24 11:09:02.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ddf46ed6bc9f658da02709e3a92dd20e6900e9bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 11:55:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/51/655151/3/check/nodejs10-npm-run-lint/ac43d15/ : SUCCESS in 4m 09s (non-voting)\n- nodejs10-npm-run-test http://logs.openstack.org/51/655151/3/check/nodejs10-npm-run-test/91df733/ : FAILURE in 5m 17s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/51/655151/3/check/openstack-tox-pep8/33b313b/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/51/655151/3/check/openstack-tox-py27/ec560f7/ : SUCCESS in 4m 06s\n- openstack-tox-py36 http://logs.openstack.org/51/655151/3/check/openstack-tox-py36/8dcaf48/ : SUCCESS in 4m 11s\n- openstack-tox-py37 http://logs.openstack.org/51/655151/3/check/openstack-tox-py37/15743e3/ : SUCCESS in 4m 24s\n- openstack-tox-docs http://logs.openstack.org/51/655151/3/check/openstack-tox-docs/5c8677e/html/ : SUCCESS in 3m 30s\n- vitrage-dashboard-integration-tests http://logs.openstack.org/51/655151/3/check/vitrage-dashboard-integration-tests/8967527/ : SUCCESS in 41m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"81e42dde336659b5b327cf80a83005181737ff31","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2019-04-24 12:01:52.000000000","message":"Patch Set 3:\n\nHi, Patch Set 3 was rebased on top of [1], since the \"Depends-on\" on commit message was not enough for the workshop in Denver.\n\n[1] https://review.opendev.org/#/c/652657/","accounts_in_message":[],"_revision_number":3},{"id":"e4dcd81c08278d88aceb3a88b9d04bdf0866fb71","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-05-12 07:00:43.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"78f7fafc24bbbeb19f8db75792a3f053da6cfc9c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-05-12 09:26:18.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"296e21a91f9463337028a2be7a1e8102bf50844a","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-05-23 13:34:40.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"363ebb9da2e632e1246ddf739868cfd28be7c2e0","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-18 06:59:20.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"35c2b06d2f04038f08b12755d305343a0456a373","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-08-11 12:39:58.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d08f69303f56ace3a4ec19382f6587cc886093c0","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-24 11:15:38.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"7263397df68f19d13ec5e2d0988f92a7fae2817a","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-24 11:27:50.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c4748b749415e1c3f09d2c28e3299fdc5a259a26","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-24 11:37:05.000000000","message":"Patch Set 4: -Workflow","accounts_in_message":[],"_revision_number":4},{"id":"cc7523a66c3f3ded1842264a8ec1af8368183b98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 11:59:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/16f79e0ee4a54f10ba501c44d5b79254 : SUCCESS in 6m 16s (non-voting)\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/bdfbb96a95c74375a2a6985b342efe31 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b1d59647fed4ed9ac6c1093f32273a7 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0aea185a2ecd4c5b99b7e8844d05fd2a : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a3dbb3ad7f2441db477dfa453bccad2 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8737dedc6d54b58aab079ef412424b0 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1a437ce61104c25986afc2da7c2ff03 : SUCCESS in 7m 08s\n- vitrage-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/fd568cb5dda94ba39d38227b8cb537db : SUCCESS in 41m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a32e54cf7c3e1eb122388c5fb172d1a89e06f941","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-24 12:04:46.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"03d63253d6d47016c3727e097289d5a074085f93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 12:04:58.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"750305cd39da8dd6c3a01018fb3841f4a51e1e9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 12:12:17.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/44369e3caf464bdcb4dc5471438592e6 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/350a84779df94040bf31738ff0cc3e27 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62bdb2cfb23a496baed0c542a9217762 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37046efddad44f0eb374a27975632579 : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":4},{"id":"b461ebcb890addf9f9c59504498f2fdb60ec56b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 12:12:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"2c5c8d3beec9144980553505266a9d8f4911d835","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 12:13:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c61ace5390b44bfd96c45c05a5f061a9 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"80bf67779b84a1d839e8c0df2ec78e6991725182","revisions":{"799629d4f6997725a791e301c05ca3addaa6ec4e":{"kind":"REWORK","_number":1,"created":"2019-04-23 13:31:45.000000000","uploader":{"_account_id":29793,"name":"Noor Yameen","email":"noor.yameen@exalt.ps","username":"NoorYameen"},"ref":"refs/changes/51/655151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/51/655151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/51/655151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/51/655151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/51/655151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/51/655151/1"}}},"commit":{"parents":[{"commit":"a98fbf3c43b165963e9b1faae14f703ae8be682c","subject":"Merge \"Changes for url transition to OpenDev\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/a98fbf3c43b165963e9b1faae14f703ae8be682c"}]}],"author":{"name":"NoorYameen","email":"noor.yameen@exalt.ps","date":"2019-04-23 13:31:33.000000000","tz":180},"committer":{"name":"NoorYameen","email":"noor.yameen@exalt.ps","date":"2019-04-23 13:32:32.000000000","tz":180},"subject":"fix alarm rca view","message":"fix alarm rca view\n\nChange-Id: If2152bd721e81cd5fdfa25af1b94e820f56c9f84\nstory: 2005502\ntask: 30604\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/799629d4f6997725a791e301c05ca3addaa6ec4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/799629d4f6997725a791e301c05ca3addaa6ec4e"}]},"branch":"refs/heads/master"},"faac1e8be223124253879e005e151dcde428ed41":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-04-23 14:09:20.000000000","uploader":{"_account_id":29793,"name":"Noor Yameen","email":"noor.yameen@exalt.ps","username":"NoorYameen"},"ref":"refs/changes/51/655151/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/51/655151/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/51/655151/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/51/655151/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/51/655151/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/51/655151/2"}}},"commit":{"parents":[{"commit":"a98fbf3c43b165963e9b1faae14f703ae8be682c","subject":"Merge \"Changes for url transition to OpenDev\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/a98fbf3c43b165963e9b1faae14f703ae8be682c"}]}],"author":{"name":"NoorYameen","email":"noor.yameen@exalt.ps","date":"2019-04-23 13:31:33.000000000","tz":180},"committer":{"name":"Noor Yameen","email":"noor.yameen@exalt.ps","date":"2019-04-23 14:09:10.000000000","tz":0},"subject":"fix alarm rca view","message":"fix alarm rca view\n\nChange-Id: If2152bd721e81cd5fdfa25af1b94e820f56c9f84\nstory: 2005502\ntask: 30604\nDepends-On: I4044338f7c62016c1fc10756fbf8ffc8485c34b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/faac1e8be223124253879e005e151dcde428ed41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/faac1e8be223124253879e005e151dcde428ed41"}]},"branch":"refs/heads/master"},"f82c40bf7e17849840e917c3c459e7382ab9da2a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-04-24 11:09:02.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/51/655151/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/51/655151/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/51/655151/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/51/655151/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/51/655151/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/51/655151/3"}}},"commit":{"parents":[{"commit":"16d145b95e5d8992638e60af9cde29f76bc75136","subject":"WIP. Update grapblib to v2.1.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/16d145b95e5d8992638e60af9cde29f76bc75136"}]}],"author":{"name":"NoorYameen","email":"noor.yameen@exalt.ps","date":"2019-04-23 13:31:33.000000000","tz":180},"committer":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2019-04-24 11:09:02.000000000","tz":0},"subject":"fix alarm rca view","message":"fix alarm rca view\n\nChange-Id: If2152bd721e81cd5fdfa25af1b94e820f56c9f84\nstory: 2005502\ntask: 30604\nDepends-On: I4044338f7c62016c1fc10756fbf8ffc8485c34b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/f82c40bf7e17849840e917c3c459e7382ab9da2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/f82c40bf7e17849840e917c3c459e7382ab9da2a"}]},"branch":"refs/heads/master"},"80bf67779b84a1d839e8c0df2ec78e6991725182":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-04-24 11:15:38.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/51/655151/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/51/655151/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/51/655151/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/51/655151/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/51/655151/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/51/655151/4"}}},"commit":{"parents":[{"commit":"5dce07ad12722b5851fd5738b0205913f05285a6","subject":"Update grapblib to v2.1.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/5dce07ad12722b5851fd5738b0205913f05285a6"}]}],"author":{"name":"NoorYameen","email":"noor.yameen@exalt.ps","date":"2019-04-23 13:31:33.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-04-24 11:15:38.000000000","tz":0},"subject":"fix alarm rca view","message":"fix alarm rca view\n\nChange-Id: If2152bd721e81cd5fdfa25af1b94e820f56c9f84\nstory: 2005502\ntask: 30604\nDepends-On: I4044338f7c62016c1fc10756fbf8ffc8485c34b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/80bf67779b84a1d839e8c0df2ec78e6991725182"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/80bf67779b84a1d839e8c0df2ec78e6991725182"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
