)]}'
{"id":"openstack%2Fvitrage-dashboard~656540","triplet_id":"openstack%2Fvitrage-dashboard~master~I36d847857c6ba162ae82679af5ea39f4e598dfd5","project":"openstack/vitrage-dashboard","branch":"master","topic":"remove-version","hashtags":[],"change_id":"I36d847857c6ba162ae82679af5ea39f4e598dfd5","subject":"Remove unused version modules","status":"MERGED","created":"2019-04-30 16:21:07.000000000","updated":"2019-05-12 07:33:12.000000000","submitted":"2019-05-12 07:33:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656540-1557646392432-0f9a02b3","meta_rev_id":"6dcc53de7bdbd16431881c5f935a21cc87b81f56","_number":656540,"virtual_id_number":656540,"owner":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"value":0,"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":2,"date":"2019-05-12 07:33:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"recommended":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"all":[{"value":1,"date":"2019-05-07 08:56:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"value":2,"date":"2019-04-30 17:28:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},{"value":0,"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-12 07:00:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-30 17:28:47.000000000","updated_by":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"reviewer":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"state":"REVIEWER"},{"updated":"2019-05-07 08:56:30.000000000","updated_by":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"reviewer":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"state":"REVIEWER"},{"updated":"2019-05-12 07:00:10.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-05-12 07:33:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d93d76680c9b3d90d09c6051782dd801b1f09b5","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-04-30 16:21:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74c91aedd44bed179f95bb5b82280582682c1a48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 17:09:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/40/656540/1/check/nodejs10-npm-run-lint/d85d273/ : SUCCESS in 3m 50s (non-voting)\n- nodejs10-npm-run-test http://logs.openstack.org/40/656540/1/check/nodejs10-npm-run-test/173d617/ : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/40/656540/1/check/openstack-tox-pep8/17bd064/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/40/656540/1/check/openstack-tox-py27/07c670b/ : SUCCESS in 4m 05s\n- openstack-tox-py36 http://logs.openstack.org/40/656540/1/check/openstack-tox-py36/36c1088/ : SUCCESS in 4m 05s\n- openstack-tox-py37 http://logs.openstack.org/40/656540/1/check/openstack-tox-py37/c573951/ : SUCCESS in 4m 20s\n- openstack-tox-docs http://logs.openstack.org/40/656540/1/check/openstack-tox-docs/29f86b5/html/ : SUCCESS in 3m 26s\n- vitrage-dashboard-integration-tests http://logs.openstack.org/40/656540/1/check/vitrage-dashboard-integration-tests/37b5246/ : SUCCESS in 43m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d9126f8ba29c597fe9cd6bb5aca973140a471402","author":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"date":"2019-04-30 17:28:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5c294bee002c9c5068e39dac68f925bca9b04c91","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2019-05-07 08:56:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"edd8b1f9ac9b099033a8e314702a1a2d9de4384a","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-05-12 07:00:10.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"da4323959dac2e8639757e1e23be5b7154ab228e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 07:00:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8f9fd00b08bf1f4f7a9e88c6b62e1579c8117b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 07:33:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/40/656540/1/gate/openstack-tox-pep8/57210db/ : SUCCESS in 4m 24s\n- openstack-tox-py27 http://logs.openstack.org/40/656540/1/gate/openstack-tox-py27/339a0a7/ : SUCCESS in 4m 18s\n- openstack-tox-py36 http://logs.openstack.org/40/656540/1/gate/openstack-tox-py36/8e0da2f/ : SUCCESS in 4m 13s\n- openstack-tox-py37 http://logs.openstack.org/40/656540/1/gate/openstack-tox-py37/03e660d/ : SUCCESS in 3m 45s\n- openstack-tox-docs http://logs.openstack.org/40/656540/1/gate/openstack-tox-docs/79a8321/html/ : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":1},{"id":"ecbbf1708238cd0a305ce3465ae6a077265eadb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 07:33:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"168bb73157e0b052654b2dc274e7d4ef36cc226c","revisions":{"168bb73157e0b052654b2dc274e7d4ef36cc226c":{"kind":"REWORK","_number":1,"created":"2019-04-30 16:21:07.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/40/656540/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/40/656540/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/40/656540/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/40/656540/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/40/656540/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/40/656540/1"}}},"commit":{"parents":[{"commit":"a98fbf3c43b165963e9b1faae14f703ae8be682c","subject":"Merge \"Changes for url transition to OpenDev\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/a98fbf3c43b165963e9b1faae14f703ae8be682c"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-04-30 16:20:26.000000000","tz":-360},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2019-04-30 16:20:26.000000000","tz":-360},"subject":"Remove unused version modules","message":"Remove unused version modules\n\nThese modules are not used anywhere and is nont working.\n\nChange-Id: I36d847857c6ba162ae82679af5ea39f4e598dfd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/168bb73157e0b052654b2dc274e7d4ef36cc226c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/168bb73157e0b052654b2dc274e7d4ef36cc226c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
