)]}'
{"id":"openstack%2Fvitrage-dashboard~662910","triplet_id":"openstack%2Fvitrage-dashboard~master~I655ee24030b06c25ded9df6e88e44bd5f4b78b6a","project":"openstack/vitrage-dashboard","branch":"master","hashtags":[],"change_id":"I655ee24030b06c25ded9df6e88e44bd5f4b78b6a","subject":"Sync Sphinx requirement","status":"MERGED","created":"2019-06-04 02:51:22.000000000","updated":"2019-06-04 09:57:28.000000000","submitted":"2019-06-04 09:57:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662910-1559642248707-7055116c","meta_rev_id":"eab0e2383becd81122b1e14ee19938add1621865","_number":662910,"virtual_id_number":662910,"owner":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":2,"date":"2019-06-04 09:57:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"all":[{"value":2,"date":"2019-06-04 09:33:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-04 06:23:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"all":[{"value":1,"date":"2019-06-04 09:33:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-04 06:23:57.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-06-04 09:33:24.000000000","updated_by":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"reviewer":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"state":"REVIEWER"},{"updated":"2019-06-04 09:57:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ff881267dc804b4b1b7b6bf8079a461954237c94","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-06-04 02:51:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a5d2f3daaeb1255bcc8811af129a0ea59bf4636","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 03:47:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/10/662910/1/check/nodejs10-npm-run-lint/8fd6be6/ : SUCCESS in 5m 46s (non-voting)\n- nodejs10-npm-run-test http://logs.openstack.org/10/662910/1/check/nodejs10-npm-run-test/964488d/ : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/10/662910/1/check/openstack-tox-pep8/af7e4b8/ : SUCCESS in 4m 50s\n- openstack-tox-py27 http://logs.openstack.org/10/662910/1/check/openstack-tox-py27/30832ba/ : SUCCESS in 4m 50s\n- openstack-tox-py36 http://logs.openstack.org/10/662910/1/check/openstack-tox-py36/c42b09e/ : SUCCESS in 3m 53s\n- openstack-tox-py37 http://logs.openstack.org/10/662910/1/check/openstack-tox-py37/50b9d68/ : SUCCESS in 4m 40s\n- openstack-tox-docs http://logs.openstack.org/10/662910/1/check/openstack-tox-docs/9cacc68/html/ : SUCCESS in 2m 37s\n- requirements-check http://logs.openstack.org/10/662910/1/check/requirements-check/1a03108/ : SUCCESS in 1m 53s\n- build-openstack-releasenotes http://logs.openstack.org/10/662910/1/check/build-openstack-releasenotes/2004931/html/ : SUCCESS in 2m 33s\n- vitrage-dashboard-integration-tests http://logs.openstack.org/10/662910/1/check/vitrage-dashboard-integration-tests/bafff51/ : SUCCESS in 54m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1127d9fc381e5dcd1480cf68b7ca52f5ff03f4a2","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-06-04 06:23:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2250bf9f3b75a72fcbe70683deda45d4d47c016c","author":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"date":"2019-06-04 09:33:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a9abd9f0cacaa6a43943948fe1e5fca4f358e418","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 09:33:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5004351c89e4dc74aa74c8c59682271d173c9a32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 09:57:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"eab0e2383becd81122b1e14ee19938add1621865","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 09:57:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/662910/1/gate/openstack-tox-pep8/ce3bd8a/ : SUCCESS in 6m 00s\n- openstack-tox-py27 http://logs.openstack.org/10/662910/1/gate/openstack-tox-py27/143ce8c/ : SUCCESS in 3m 56s\n- openstack-tox-py36 http://logs.openstack.org/10/662910/1/gate/openstack-tox-py36/14229e9/ : SUCCESS in 3m 39s\n- openstack-tox-py37 http://logs.openstack.org/10/662910/1/gate/openstack-tox-py37/32eb4b2/ : SUCCESS in 6m 18s\n- openstack-tox-docs http://logs.openstack.org/10/662910/1/gate/openstack-tox-docs/a676d0d/html/ : SUCCESS in 3m 55s\n- requirements-check http://logs.openstack.org/10/662910/1/gate/requirements-check/331f9ed/ : SUCCESS in 2m 14s\n- build-openstack-releasenotes http://logs.openstack.org/10/662910/1/gate/build-openstack-releasenotes/390cfb2/html/ : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"76a88798154e0326f817566fa9ab6bc0f8ff7446","revisions":{"76a88798154e0326f817566fa9ab6bc0f8ff7446":{"kind":"REWORK","_number":1,"created":"2019-06-04 02:51:22.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/10/662910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/10/662910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/10/662910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/10/662910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/10/662910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/10/662910/1"}}},"commit":{"parents":[{"commit":"168bb73157e0b052654b2dc274e7d4ef36cc226c","subject":"Remove unused version modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/168bb73157e0b052654b2dc274e7d4ef36cc226c"}]}],"author":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-06-04 02:50:26.000000000","tz":480},"committer":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-06-04 02:50:57.000000000","tz":480},"subject":"Sync Sphinx requirement","message":"Sync Sphinx requirement\n\nSync sphinx dependency with global requirements. It caps python 2 since\nsphinx 2.0 no longer supports Python 2.7.\n\nChange-Id: I655ee24030b06c25ded9df6e88e44bd5f4b78b6a\nCloses-Bug: #1831528\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/76a88798154e0326f817566fa9ab6bc0f8ff7446"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/76a88798154e0326f817566fa9ab6bc0f8ff7446"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
