)]}'
{"id":"openstack%2Fvitrage-dashboard~669022","triplet_id":"openstack%2Fvitrage-dashboard~master~Ic509bdfc23d95d709db8b534827ab141e3d2a0b0","project":"openstack/vitrage-dashboard","branch":"master","topic":"python3-train","hashtags":[],"change_id":"Ic509bdfc23d95d709db8b534827ab141e3d2a0b0","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-04 02:21:45.000000000","updated":"2019-07-04 12:36:41.000000000","submitted":"2019-07-04 12:36:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669022-1562243801043-4f1731f4","meta_rev_id":"3d056a4874b8e2c2d4ad19f8f46c0399c44a2e8f","_number":669022,"virtual_id_number":669022,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":2,"date":"2019-07-04 12:36:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"all":[{"value":2,"date":"2019-07-04 12:26:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-04 11:22:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"all":[{"value":1,"date":"2019-07-04 12:26:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 11:22:24.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-07-04 12:26:51.000000000","updated_by":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"reviewer":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"state":"REVIEWER"},{"updated":"2019-07-04 12:36:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d016662010a8c88bcfceb3c107cbb77081aa264","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-04 02:21:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4ab8ce0ea85b4c13afb0fd19835c3c76901b449","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 03:19:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/22/669022/1/check/nodejs10-npm-run-lint/e28408d/ : SUCCESS in 4m 30s (non-voting)\n- nodejs10-npm-run-test http://logs.openstack.org/22/669022/1/check/nodejs10-npm-run-test/f73afb9/ : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/22/669022/1/check/openstack-tox-pep8/0da6240/ : SUCCESS in 4m 56s\n- openstack-tox-py27 http://logs.openstack.org/22/669022/1/check/openstack-tox-py27/6b61811/ : SUCCESS in 4m 42s\n- openstack-tox-py36 http://logs.openstack.org/22/669022/1/check/openstack-tox-py36/013edcf/ : SUCCESS in 3m 13s\n- openstack-tox-py37 http://logs.openstack.org/22/669022/1/check/openstack-tox-py37/7ac224b/ : SUCCESS in 3m 34s\n- openstack-tox-docs http://logs.openstack.org/22/669022/1/check/openstack-tox-docs/49870f8/html/ : SUCCESS in 3m 45s\n- vitrage-dashboard-integration-tests http://logs.openstack.org/22/669022/1/check/vitrage-dashboard-integration-tests/77fcbf4/ : SUCCESS in 49m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7364b179b6c0fa69d128cbb0c9dd26ce308ac34d","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-07-04 11:22:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"56831f90731b1f740157e606133a715d97fa4c88","author":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"date":"2019-07-04 12:26:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9d18c802783cf24ade754f3f0c14cccde4c6c849","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 12:27:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f0f79b8ff33b107f05e755dbbbdfae579dfaacf9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 12:36:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/22/669022/1/gate/openstack-tox-pep8/6afc437/ : SUCCESS in 4m 14s\n- openstack-tox-py27 http://logs.openstack.org/22/669022/1/gate/openstack-tox-py27/a0e6705/ : SUCCESS in 3m 41s\n- openstack-tox-py36 http://logs.openstack.org/22/669022/1/gate/openstack-tox-py36/13c0d11/ : SUCCESS in 3m 32s\n- openstack-tox-py37 http://logs.openstack.org/22/669022/1/gate/openstack-tox-py37/1300589/ : SUCCESS in 4m 02s\n- openstack-tox-docs http://logs.openstack.org/22/669022/1/gate/openstack-tox-docs/d4fde5d/html/ : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"428a5953b00350be41fb4a80f685215faae35028","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 12:36:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"23cbc09b4dd047a93e81be6185d41eb8118c5ad4","revisions":{"23cbc09b4dd047a93e81be6185d41eb8118c5ad4":{"kind":"REWORK","_number":1,"created":"2019-07-04 02:21:45.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/22/669022/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/22/669022/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/22/669022/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/22/669022/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/22/669022/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/22/669022/1"}}},"commit":{"parents":[{"commit":"1f39f50f4a083fa8444c2e7d0b20c5d5f7c5ad80","subject":"Update the constraints url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/1f39f50f4a083fa8444c2e7d0b20c5d5f7c5ad80"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-04 02:21:01.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-04 02:21:46.000000000","tz":480},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: Ic509bdfc23d95d709db8b534827ab141e3d2a0b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/23cbc09b4dd047a93e81be6185d41eb8118c5ad4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/23cbc09b4dd047a93e81be6185d41eb8118c5ad4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
