)]}'
{"id":"openstack%2Fvitrage-dashboard~669526","triplet_id":"openstack%2Fvitrage-dashboard~master~I533d0459f74230e551ddd38f40c3bf4c568e3c21","project":"openstack/vitrage-dashboard","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I533d0459f74230e551ddd38f40c3bf4c568e3c21","subject":"Limit envlist to py37 for Python 3 Train goal","status":"MERGED","created":"2019-07-05 20:45:02.000000000","updated":"2019-07-22 12:32:30.000000000","submitted":"2019-07-22 12:32:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669526-1563798750608-a91269ad","meta_rev_id":"d1d059fa27e8fa0a370f3994c9d37fc5fa22d24c","_number":669526,"virtual_id_number":669526,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2019-07-22 12:32:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2019-07-22 12:14:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-12 06:45:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":1,"date":"2019-07-22 12:19:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 10:48:28.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2019-07-12 06:45:39.000000000","updated_by":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"reviewer":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"state":"REVIEWER"},{"updated":"2019-07-22 12:14:34.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-07-22 12:19:18.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-07-22 12:32:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"58745488e58483ef8ba1abb5d9ac9ec101b755bd","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 20:45:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81c2ac1642bc368de85114048ed82df8e494bcc2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 21:32:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/26/669526/1/check/nodejs10-npm-run-lint/19ea73f/ : SUCCESS in 5m 22s (non-voting)\n- nodejs10-npm-run-test http://logs.openstack.org/26/669526/1/check/nodejs10-npm-run-test/1bf2f2d/ : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/26/669526/1/check/openstack-tox-pep8/3f59233/ : SUCCESS in 3m 41s\n- openstack-tox-py27 http://logs.openstack.org/26/669526/1/check/openstack-tox-py27/e7b3790/ : SUCCESS in 4m 09s\n- openstack-tox-py36 http://logs.openstack.org/26/669526/1/check/openstack-tox-py36/f0ad6f6/ : SUCCESS in 3m 42s\n- openstack-tox-py37 http://logs.openstack.org/26/669526/1/check/openstack-tox-py37/276f2fb/ : SUCCESS in 3m 45s\n- openstack-tox-docs http://logs.openstack.org/26/669526/1/check/openstack-tox-docs/88100e2/html/ : SUCCESS in 3m 11s\n- build-openstack-releasenotes http://logs.openstack.org/26/669526/1/check/build-openstack-releasenotes/d3a62aa/html/ : SUCCESS in 2m 51s\n- vitrage-dashboard-integration-tests http://logs.openstack.org/26/669526/1/check/vitrage-dashboard-integration-tests/61b1101/ : SUCCESS in 39m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b05570d212761105e4f3c0f03f14b139c081068b","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-07-11 10:48:28.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3a04af4ec5e3625c15a96efb0891f773d72692f","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-11 16:34:25.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2b92a39391f9865c31b277c4b6175bd21e3461d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 17:14:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/26/669526/2/check/nodejs10-npm-run-lint/8cf1017/ : SUCCESS in 5m 35s (non-voting)\n- nodejs10-npm-run-test http://logs.openstack.org/26/669526/2/check/nodejs10-npm-run-test/cc23c4d/ : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/26/669526/2/check/openstack-tox-pep8/d4853d2/ : SUCCESS in 3m 55s\n- openstack-tox-py27 http://logs.openstack.org/26/669526/2/check/openstack-tox-py27/e0b4a43/ : SUCCESS in 4m 30s\n- openstack-tox-py36 http://logs.openstack.org/26/669526/2/check/openstack-tox-py36/c2bac06/ : SUCCESS in 3m 20s\n- openstack-tox-py37 http://logs.openstack.org/26/669526/2/check/openstack-tox-py37/c819cd0/ : SUCCESS in 4m 36s\n- openstack-tox-docs http://logs.openstack.org/26/669526/2/check/openstack-tox-docs/3668e01/html/ : SUCCESS in 3m 24s\n- build-openstack-releasenotes http://logs.openstack.org/26/669526/2/check/build-openstack-releasenotes/58e307e/html/ : SUCCESS in 3m 43s\n- vitrage-dashboard-integration-tests http://logs.openstack.org/26/669526/2/check/vitrage-dashboard-integration-tests/729c923/ : SUCCESS in 37m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"14c6f47deb83a6f67603c5c7f266d634e6f705de","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-12 06:45:39.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"562017709131e6adb7a94e8d736c3e002b317074","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-07-22 12:14:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"84c0f5786affb0087a531052b06cd6d3cfdfcbe4","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-07-22 12:19:18.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"df8a2811fd3850a7dfdd5ca6c489f7b10e18aa18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 12:19:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ff3719ef253a39a8a8922db4bdb638f0961c27ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 12:32:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/26/669526/2/gate/openstack-tox-pep8/9091225/ : SUCCESS in 4m 49s\n- openstack-tox-py27 http://logs.openstack.org/26/669526/2/gate/openstack-tox-py27/e9f9a02/ : SUCCESS in 7m 05s\n- openstack-tox-py36 http://logs.openstack.org/26/669526/2/gate/openstack-tox-py36/5da0c9f/ : SUCCESS in 3m 25s\n- openstack-tox-py37 http://logs.openstack.org/26/669526/2/gate/openstack-tox-py37/6c8f9f3/ : SUCCESS in 4m 00s\n- openstack-tox-docs http://logs.openstack.org/26/669526/2/gate/openstack-tox-docs/25a231c/html/ : SUCCESS in 3m 08s\n- build-openstack-releasenotes http://logs.openstack.org/26/669526/2/gate/build-openstack-releasenotes/394577f/html/ : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"e33fbf84ef90171e7a7bb617a70743548b3b554c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 12:32:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f66683c01c98b1575342784d1ab2143b3ac87b72","revisions":{"a1c2ff60db894dec93e8ead72416154d27bf3d08":{"kind":"REWORK","_number":1,"created":"2019-07-05 20:45:02.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/26/669526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/26/669526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/26/669526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/26/669526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/26/669526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/26/669526/1"}}},"commit":{"parents":[{"commit":"d9b8cfd1e8202e6bf077d6889d6800d042b16a72","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/d9b8cfd1e8202e6bf077d6889d6800d042b16a72"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 16:08:57.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 20:45:01.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I533d0459f74230e551ddd38f40c3bf4c568e3c21\nStory: #2005924\nTask: #34255\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/a1c2ff60db894dec93e8ead72416154d27bf3d08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/a1c2ff60db894dec93e8ead72416154d27bf3d08"}]},"branch":"refs/heads/master"},"f66683c01c98b1575342784d1ab2143b3ac87b72":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-11 16:34:25.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/26/669526/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/26/669526/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/26/669526/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/26/669526/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/26/669526/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/26/669526/2"}}},"commit":{"parents":[{"commit":"d9b8cfd1e8202e6bf077d6889d6800d042b16a72","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/d9b8cfd1e8202e6bf077d6889d6800d042b16a72"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 16:08:57.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-11 16:34:09.000000000","tz":-240},"subject":"Limit envlist to py37 for Python 3 Train goal","message":"Limit envlist to py37 for Python 3 Train goal\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train. The correct zuul jobs are\nalready in place so this change just limits default \u0027tox\u0027 calls to py37\nto save developers time.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I533d0459f74230e551ddd38f40c3bf4c568e3c21\nStory: #2005924\nTask: #34255\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/f66683c01c98b1575342784d1ab2143b3ac87b72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/f66683c01c98b1575342784d1ab2143b3ac87b72"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
