)]}'
{"id":"openstack%2Fvitrage-dashboard~700940","triplet_id":"openstack%2Fvitrage-dashboard~master~I7dd429b23f2d28999dc30e13d3717d8087f202aa","project":"openstack/vitrage-dashboard","branch":"master","topic":"drop-django111-support","hashtags":[],"change_id":"I7dd429b23f2d28999dc30e13d3717d8087f202aa","subject":"Drop Django 1.11 support","status":"MERGED","created":"2020-01-02 20:53:15.000000000","updated":"2020-01-07 15:11:17.000000000","submitted":"2020-01-07 15:09:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700940-1578409760460-b2b5e395","meta_rev_id":"cefd52de15c8958fb173289a2f4f86db2f86774d","_number":700940,"virtual_id_number":700940,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-07 15:09:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-07 14:09:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-07 15:02:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-07 15:02:06.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-01-07 15:09:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6d7d59c2375642799ff49b3c5748932a8342e9f1","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-02 20:53:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b96ff803b0b1b19a7308f30abf8c4399078a75c","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-02 20:53:28.000000000","message":"Topic set to drop-django111-support","accounts_in_message":[],"_revision_number":1},{"id":"298108bd66db23811385fa8f2e86442ad6fc6b84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 21:38:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/fe67c4a7d58e445f8b04082ef4375027 : SUCCESS in 5m 15s (non-voting)\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/741dd726f33d49fb82e7a0bdce30f597 : SUCCESS in 4m 56s (non-voting)\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/9e211e0be9cb40f1a6f48d7a412de3d2 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a806c9077d5247988020d23d076503aa : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb712ea6b7c4421783d2f7a387905ee9 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2635ba7c7a59477798d05aae18ca4504 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7595fa4ba004408a83f96ec06feda42b : SUCCESS in 4m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b84c99c1239c4f1590451180fb727a60 : SUCCESS in 2m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9b843d1997643f088d06780ab65a96f : SUCCESS in 3m 14s\n- vitrage-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/6a5f4113e7894105af8997660263b905 : SUCCESS in 39m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"944c32d3674676cbbfb3b73bc62f47aedf09f6cb","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-07 07:45:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fe6441c99ae2f1dd642f45404a03dddbe60b754f","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-07 14:08:52.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4caf9b2706d2f0c5d55213a82771bcd98d84efa3","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-07 14:09:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c2ca7be1c4b545d995446ecbd31b7f27ca78f5c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 14:49:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/84dfbb471f854c14bf0703572405ad7d : SUCCESS in 6m 16s (non-voting)\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/71ade30a6e584c339935e1eeb1531178 : SUCCESS in 6m 44s (non-voting)\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/1206ff5554054e3ca0563b247cd1cf4e : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d9be4f56244469093987ce38e59ecba : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53426982d45d4e4480a99577dec40ed6 : SUCCESS in 7m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/502e5b5d28f84f65a568106f8d63ced3 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15b48124b59b45759a6a82c90e738314 : SUCCESS in 7m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1dc81020429f4caaa7147a043243d463 : SUCCESS in 3m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e0102504b8943c8a20caf271cfdfe97 : SUCCESS in 3m 57s\n- vitrage-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/bf6a83838aff46e5a1a69cd825dcd0a0 : SUCCESS in 38m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9ffbe556946f53ea52926f469799d75b8b691512","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-07 15:02:06.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"198e394bb2ffdeb05f0e21695fbbe7a2063884ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 15:02:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7b92369549ff3d6ab5eb2aa1af2e1030b5d88536","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 15:09:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"13fb9d6adfd54976e3cc126c2f202645027c69cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 15:09:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/58ce59bd6aa940cda50611c605159362 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf2057d83deb42f3b21f863961c58abb : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfe9d71b5db74abb8597e4d28d16f499 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f84f44049f3945a3a40c4ab56a744e0c : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2de47cf22624384bdd4cdc14ada24d6 : SUCCESS in 6m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b47648b0bc364c3e83ef8946645677a5 : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce06bb125d06470b89d8440c1d64b96c : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":2},{"id":"cefd52de15c8958fb173289a2f4f86db2f86774d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 15:11:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6a4dee1c415491a91029c55e2de4ec5 : SUCCESS in 1m 33s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12d074abefbd4542bd1b9bdb245ffb15 : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"938fa9aab4b2118b76340a910475b1fa1a027a3b","revisions":{"a8e92d0c39bff773c9fe77a6dc1618a978f8e577":{"kind":"REWORK","_number":1,"created":"2020-01-02 20:53:15.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/40/700940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/40/700940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/40/700940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/40/700940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/40/700940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/40/700940/1"}}},"commit":{"parents":[{"commit":"96f8abebcc7b7c69af920f0736861980330a8256","subject":"Merge \"Change README.rst with a better title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/96f8abebcc7b7c69af920f0736861980330a8256"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-02 20:52:33.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-02 20:53:13.000000000","tz":0},"subject":"Drop Django 1.11 support","message":"Drop Django 1.11 support\n\nDjango 1.11 ends its extended support in April 2020 (which is before\nUssuri release), so horizon drops Django 1.11 support in Ussuri.\n\ntox envs for non-primary Django versions are no longer needed in tox.ini\nas testing environments for non-primary Django versions are setup in\nthe zuul jobs now.\n\nhorizon\u003e\u003d17.1.0 is required to use Django 2.2.\nrequirements.txt is updated accordingly.\n\nDepends-On: https://review.opendev.org/#/c/700733/\nChange-Id: I7dd429b23f2d28999dc30e13d3717d8087f202aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/a8e92d0c39bff773c9fe77a6dc1618a978f8e577"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/a8e92d0c39bff773c9fe77a6dc1618a978f8e577"}]},"branch":"refs/heads/master"},"938fa9aab4b2118b76340a910475b1fa1a027a3b":{"kind":"REWORK","_number":2,"created":"2020-01-07 14:08:52.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/40/700940/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/40/700940/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/40/700940/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/40/700940/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/40/700940/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/40/700940/2"}}},"commit":{"parents":[{"commit":"96f8abebcc7b7c69af920f0736861980330a8256","subject":"Merge \"Change README.rst with a better title\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/96f8abebcc7b7c69af920f0736861980330a8256"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-01-02 20:52:33.000000000","tz":0},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-07 14:08:49.000000000","tz":0},"subject":"Drop Django 1.11 support","message":"Drop Django 1.11 support\n\nDjango 1.11 ends its extended support in April 2020 (which is before\nUssuri release), so horizon drops Django 1.11 support in Ussuri.\n\ntox envs for non-primary Django versions are no longer needed in tox.ini\nas testing environments for non-primary Django versions are setup in\nthe zuul jobs now.\n\nhorizon\u003e\u003d17.1.0 is required to use Django 2.2.\nrequirements.txt is updated accordingly.\n\nDepends-On: https://review.opendev.org/#/c/700733/\nChange-Id: I7dd429b23f2d28999dc30e13d3717d8087f202aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/938fa9aab4b2118b76340a910475b1fa1a027a3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/938fa9aab4b2118b76340a910475b1fa1a027a3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
