)]}'
{"id":"openstack%2Fvitrage-dashboard~709208","triplet_id":"openstack%2Fvitrage-dashboard~master~I07685dfbde7c96c5b8da7e93af291df16c3b8bad","project":"openstack/vitrage-dashboard","branch":"master","hashtags":[],"change_id":"I07685dfbde7c96c5b8da7e93af291df16c3b8bad","subject":"Replace embedded static files with XStatic-*","status":"MERGED","created":"2020-02-21 18:07:00.000000000","updated":"2020-04-01 10:54:53.000000000","submitted":"2020-04-01 10:50:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709208-1585738246831-93ed6149","meta_rev_id":"9e05522949350da8c0f1b60364f8ed40db998e01","_number":709208,"virtual_id_number":709208,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"date":"2020-03-27 11:20:51.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-01 10:50:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":2,"date":"2020-04-01 10:36:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-01 10:38:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-01 10:36:53.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-04-01 10:38:11.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-04-01 10:50:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"96a2b2191a3b05e02c656e8ee30912f29f22e470","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-02-21 18:07:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6edded29e09c57dafce5ee812ec4d24e6a93e87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 18:46:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/f613dbf42a204fe8b2e8003c9f3c2c1b : SUCCESS in 7m 15s (non-voting)\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/9775b198d4d143eb9b7179ab9f5d69ca : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/99742410120a45eb95736b82755b53a8 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aae3b490be4244ffa6cf79b2a0710552 : FAILURE in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef6c5e727b8741bd8c67642779d5f355 : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ed95e5147714213a5cde2174cc51916 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/388ef5f9db784e40ad0c35276477592c : SUCCESS in 8m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fd052cf6503e4e3d90056e046042780c : FAILURE in 4m 36s\n- vitrage-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/aa9d46aea2964352a52a93a4274aa41b : SUCCESS in 34m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"86269cfc3abe088ef3dac8abe21d77fe6cb1dbde","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-03-12 17:27:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5dc83d4ddf7d83b8121a7ca9fbcb48d12857f557","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 18:25:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/e39bab83387247e99b05bd0be8b4e76c : SUCCESS in 6m 11s (non-voting)\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/de7f23d6ce8e4a90bddeb2e84218ec1f : SUCCESS in 8m 38s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ca599c4feb44f10a89c42a6f9c22cd1 : FAILURE in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f1050bcce694d9d9d129403dab72586 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f026a29ec6c849bc97a7196ad1b159fd : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d43aa6639d814dc8b897e4dd203d6123 : FAILURE in 3m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2f05f96329049e7b7ca7b08bb4a6362 : SUCCESS in 6m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d2e3fcd1594744069b0dfead527c5335 : SUCCESS in 2m 09s\n- vitrage-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/af86fdb7828d4a53aeda6571b76952ba : SUCCESS in 54m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ad946cc8292c139b5f119f53aec2aa1690957743","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-03-12 19:11:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"769c7b4b93fea547171454166386c7bfb051c455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 20:02:07.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/eecd25c0ba8842e7886ffa013fcd1a56 : SUCCESS in 5m 19s (non-voting)\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/8b67f1defa8546fe8ac0ca8b9d010d17 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f0db5f35fd3b4ee2970a05d474884d13 : FAILURE in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08621e9788814be5b61eff874fb05322 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f9eba1b78974c1abc2e85fd0fd83be7 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ab850e33d37476c8e7a47b83913398f : FAILURE in 3m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c21f32a36809417785dbaf5d58e6741c : SUCCESS in 4m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bca5382c72544c95af85b964e1ebac54 : SUCCESS in 2m 22s\n- vitrage-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/6f64bc8f76234dc4b8812d8c2820b055 : SUCCESS in 45m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6207333952fcd7fd25a187ff751805fd95a566ea","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-03-17 19:33:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d6e42133df11023582ed272d7a634b9a006ea155","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 20:25:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/6d46f7c0e51546ada090e85a2f7cec02 : SUCCESS in 8m 01s (non-voting)\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/305bf2b556d24330ba90d0fd0c95240d : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df47ef78b6f24d32bc4909263d2efeee : FAILURE in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/312a488ed32d48fab6fc05c5ff31c90b : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/163ff1fc6afa491e93536046b29989aa : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db6ebdcdfe664c97a8ee5068fc352d9a : FAILURE in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3542eb264d94bfa9474a13db07a1e45 : SUCCESS in 9m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/eea6639fa21d4e3b8ef099f068775fe2 : SUCCESS in 4m 50s\n- vitrage-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/c29e286705da443cbd75edc653b1a560 : SUCCESS in 46m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"99789b87ca130210ceeee0ad8fbd2f4ad94a5fbb","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-03-18 09:36:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fbf1a516bf9178956f61808652c7bd217b8a0189","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 10:25:31.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/d6b06006e94c453bbff5e7ac211c7cd2 : SUCCESS in 5m 25s (non-voting)\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/9b8d2ba213244ca2a3cd27ebf824c483 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22cab17537f64c22a5eebf7432eb1be5 : FAILURE in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5fa12505f614dde942883856f365d45 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29f93d8c20ae41dfb7994373748ac9e5 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/111e22142a2148da934471888c38699c : FAILURE in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed55b80b016d499ca363ddf310730d95 : SUCCESS in 5m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9b01363dc8094af9ade97731320813f9 : SUCCESS in 2m 20s\n- vitrage-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/4ee15303ad034693b2075d16ac373efb : SUCCESS in 39m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cf529153c2b46f73f0e2ac162cabc702710676b2","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-03-18 10:46:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"95f4da59421db882d8d778a70befa8772d7ce249","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 11:38:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/52e89386d04c484caa139e998a9cd54f : SUCCESS in 6m 22s (non-voting)\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/3e2466de7a41446fab14a3aebdd2636a : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6dbe7a0cfa914f28a14b3461495299ab : FAILURE in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbfd92b038fe463b99fea8f1ccd55a9c : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ff5a939562c4e93b71ec70a9539f510 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/037d4cd95f4e4229859cb5d374443c51 : FAILURE in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5c380f76dda4e08a9df95d93030f119 : SUCCESS in 5m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/58acc0f8297b4062a0232090a517dbf6 : SUCCESS in 3m 57s\n- vitrage-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/2fc6a51424f44642b820da99520781e3 : SUCCESS in 45m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"de6776d5ad0973ef955246943854aabb93b35c14","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-03-23 17:55:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"859b39c966c046b3be2fb244605b376aa51c4a35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 18:47:20.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/bf490e69ff4e4787bb689d7fe4ab5b9e : SUCCESS in 5m 52s (non-voting)\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/a9bf2e6ba87a4be4aa63e16627ab64ac : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/283fb8d08bf94dc19c2ba25c884dd442 : FAILURE in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dae1ad545db4e1bb830b3776648f9a8 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67b3fd64f7494eb0a572caf19c49e7a6 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8965682c94ac47d894d83ac31564305a : FAILURE in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/058f62d8413a473f8caa8cfa66792006 : FAILURE in 6m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3d98a1ed4a1b492d8753bd724a270f20 : SUCCESS in 2m 50s\n- vitrage-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/456c04f2873744568eeee039c7b08cce : SUCCESS in 48m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3a33abe12c82d544a0e87660d3d53caae204c1a2","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-03-24 18:42:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c5f46cb431d26a9138731fbb62095bb7695a7ae2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 20:55:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/0d5144d02aff4837b85250f49a612720 : SUCCESS in 7m 46s (non-voting)\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/0d49429653a2452896ab82501866632c : SUCCESS in 10m 14s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/23fbb366f57b4785bf035ea71276204f : FAILURE in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5f5da635e654d9cb04e3836949e17cf : SUCCESS in 8m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae16aed62fd744a0bd610d336ba8bcd9 : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40a0b8cd00e640abb83fc356647b935f : FAILURE in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f7dac2124b04153b167944bb2447df4 : FAILURE in 7m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4b99e066809f4e3d981946ced5b6ab19 : SUCCESS in 4m 44s\n- vitrage-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/74bc4d5584bf4d11a6314f5e97228f5d : SUCCESS in 46m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d24d8c51ffca92f0b6f107212476f852372436f3","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-03-25 13:05:39.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0a9045a606ec997d3fe397ac1f3f6f4857061692","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 13:47:38.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/9d8de7f2ef5f4549b4f2b876924de28b : SUCCESS in 4m 55s (non-voting)\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/f3f7d4014070451c878d62d0cf7482bd : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c212e21ca454fa5b1005650f854d136 : FAILURE in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e53199975e5c42f283e4b82023b221bc : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6ed8835267446828cc6665b1ba750c5 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a2aa040ec6b47fb9b93c3ba8f243032 : FAILURE in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d63c2ed1d0c4a53a90973eee570c96e : SUCCESS in 5m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/70c4c5c1752e4a22835b372758641bfb : SUCCESS in 2m 46s\n- vitrage-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/9461815805f54a52b3afe0df1ad9ebe6 : SUCCESS in 41m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7bd3c47e309bc7940415e162184aaf946aadfef6","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-03-27 11:20:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0e012f46bc82c7bcc43a7d68573d56b7aaf5a162","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 12:24:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/4ad604b101244e139950ff5cbbbfaf08 : SUCCESS in 6m 48s (non-voting)\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/2d764fa299614f4b98e5eba1b331f185 : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d891489e91f458897c450571ef0d3fb : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48759c087da647b4b42248c668de091a : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abbb7f664b81431581a5247e0d0b9bad : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c432f8f026164d518c864ed60419732a : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cc6904809184c2daca93b868eb3e45b : SUCCESS in 8m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a252e8978e2447bc838ecb88f03fa9a6 : SUCCESS in 4m 42s\n- vitrage-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/0e9ae969e5944816a7e652b90f29e710 : SUCCESS in 59m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d39d76b9f6f9331ba868cce0bc7ccb07601e0c7e","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-01 10:36:53.000000000","message":"Patch Set 2: Code-Review+2\n\nVerified on devstack! Thank you, Michal!","accounts_in_message":[],"_revision_number":2},{"id":"ca63a95774ca9ce934551c50d73c7b92a7c2f85d","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-01 10:38:11.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"68729bc75d9bce0de7bd937732f79ceda4e804c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 10:38:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5b6eb032505c5e4ff80bba6fc4feed8e9e516881","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 10:50:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa57d6fc16b44eb69abfed3d2ad242f5 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3ac06b6ddec4131b440128b487e9f99 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/569119a5ede044c0bcbb1ab32b472ecd : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02ef489b101a420f8493257190485944 : SUCCESS in 7m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/456357bff0914f459f86e252c612fd0d : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":2},{"id":"d9e0fa48d3b0ff17cd087db0eaf759ffe25665c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 10:50:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9e05522949350da8c0f1b60364f8ed40db998e01","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 10:54:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15196e2aaeb148fd8c2a13e9f3840b90 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d181623f2285b7d745fe161e703dea7114ec8376","revisions":{"0fb52a77279916a8e828f92e778134158c119f90":{"kind":"REWORK","_number":1,"created":"2020-02-21 18:07:00.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/08/709208/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/08/709208/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/08/709208/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/08/709208/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/08/709208/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/08/709208/1"}}},"commit":{"parents":[{"commit":"c829088a91eaa6734eed3eb2a12817e615d1ee46","subject":"Merge \"Drop Django 1.11 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/c829088a91eaa6734eed3eb2a12817e615d1ee46"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-02-21 17:56:07.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-02-21 18:06:12.000000000","tz":60},"subject":"Replace embedded static files with XStatic-*","message":"Replace embedded static files with XStatic-*\n\nRemove embedded js, add reference in settings,\nadd requirement xstatic libs.\n\nChange-Id: I07685dfbde7c96c5b8da7e93af291df16c3b8bad\nStory: #2004051\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/0fb52a77279916a8e828f92e778134158c119f90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/0fb52a77279916a8e828f92e778134158c119f90"}]},"branch":"refs/heads/master"},"d181623f2285b7d745fe161e703dea7114ec8376":{"kind":"REWORK","_number":2,"created":"2020-03-12 17:27:00.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/08/709208/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/08/709208/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/08/709208/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/08/709208/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/08/709208/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/08/709208/2"}}},"commit":{"parents":[{"commit":"c829088a91eaa6734eed3eb2a12817e615d1ee46","subject":"Merge \"Drop Django 1.11 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/c829088a91eaa6734eed3eb2a12817e615d1ee46"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-02-21 17:56:07.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2020-03-12 17:26:19.000000000","tz":60},"subject":"Replace embedded static files with XStatic-*","message":"Replace embedded static files with XStatic-*\n\nRemove embedded js, add reference in settings,\nadd requirement xstatic libs.\n\nChange-Id: I07685dfbde7c96c5b8da7e93af291df16c3b8bad\nDepends-On: https://review.opendev.org/712744\nStory: #2004051\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/d181623f2285b7d745fe161e703dea7114ec8376"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/d181623f2285b7d745fe161e703dea7114ec8376"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
