)]}'
{"id":"openstack%2Fvitrage-dashboard~744629","triplet_id":"openstack%2Fvitrage-dashboard~master~Ief055dbc611ef8e53c7bca5c8130614ff8e3ae22","project":"openstack/vitrage-dashboard","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"Ief055dbc611ef8e53c7bca5c8130614ff8e3ae22","subject":"DNM: testing gate on ubuntu focal","status":"ABANDONED","created":"2020-08-04 03:33:42.000000000","updated":"2020-08-04 15:45:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3ed0c216518a764e42cb4309aca0432f56bf1588","_number":744629,"virtual_id_number":744629,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-04 05:36:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2020-08-04 06:20:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-04 05:36:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-04 06:20:31.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"e61a176880c935ab7b3903c2733736a888d0433a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-04 03:33:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b79b6cf845e83c1515ee2f6586b5958496fc28b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 05:36:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/bf47a6f1031e4eaa87645d831fe96beb : SUCCESS in 5m 40s (non-voting)\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/ea826e07f7dd4562aeedc8485df747b2 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50927505120d4651a25b27894e58055e : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dde39bafcfe34f5ca00c9f835fcf84a2 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8739caa33a86419595d8b734d94f683f : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac83ba2462e14447b6e2a31d17232801 : SUCCESS in 5m 34s\n- vitrage-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/cf3c7bab9da04c5f89f41628cfacc253 : SUCCESS in 38m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2871094863ef85afe1ce4790fad8e33ecc4ae9b5","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-08-04 06:17:26.000000000","message":"Patch Set 1: Code-Review+1\n\nThere is no issue for focal-migration in vitrage-dashboard.\nSo let\u0027s remove the depends-on now to get this merge before base jobs switch to Focal.","accounts_in_message":[],"_revision_number":1},{"id":"d2be21a16cae3ba393db62ab48f21bc74f78c4a1","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-08-04 06:20:31.000000000","message":"Patch Set 1: -Code-Review\n\n\u003e There is no issue for focal-migration in vitrage-dashboard.\n \u003e So let\u0027s remove the depends-on now to get this merge before base\n \u003e jobs switch to Focal.\n\noops. I mean there is no change required for this repo. Let\u0027s abandon it.","accounts_in_message":[],"_revision_number":1},{"id":"1eb6d9bc6baf2f6811fa89b6909efef10a971f55","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-04 15:45:23.000000000","message":"Abandoned\n\nall working fine, no updates needed for Focal migration.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5213d2f966581081c793f017c55fab8530314940","revisions":{"5213d2f966581081c793f017c55fab8530314940":{"kind":"REWORK","_number":1,"created":"2020-08-04 03:33:42.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/29/744629/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/29/744629/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/29/744629/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/29/744629/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/29/744629/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/29/744629/1"}}},"commit":{"parents":[{"commit":"53026b9867610b313c71128ef88b24d6102bccda","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/53026b9867610b313c71128ef88b24d6102bccda"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-04 03:01:58.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-04 03:01:58.000000000","tz":-300},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40226\n\nDepends-On: https://review.opendev.org/#/c/734700/\nChange-Id: Ief055dbc611ef8e53c7bca5c8130614ff8e3ae22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/5213d2f966581081c793f017c55fab8530314940"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/5213d2f966581081c793f017c55fab8530314940"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
