)]}'
{"id":"openstack%2Fvitrage-dashboard~766993","triplet_id":"openstack%2Fvitrage-dashboard~master~Id66915db78d788d98645d536fb112dadf9a89aed","project":"openstack/vitrage-dashboard","branch":"master","topic":"Use-Phantomjs","hashtags":[],"change_id":"Id66915db78d788d98645d536fb112dadf9a89aed","subject":"Run npm nodejs job with Phantomjs browser","status":"MERGED","created":"2020-12-14 16:44:15.000000000","updated":"2020-12-15 16:02:00.000000000","submitted":"2020-12-15 15:59:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766993-Use-Phantomjs","meta_rev_id":"35ce42ea78a519c905cc40dc61c36ff5c71ded82","_number":766993,"virtual_id_number":766993,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-15 15:59:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":2,"date":"2020-12-15 15:45:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":1,"date":"2020-12-15 15:45:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-14 17:25:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-15 15:45:41.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"}],"messages":[{"id":"15a618dcdec13e845e41423750814540c80c6172","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-12-14 16:44:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd5293025efbab3e93b066200ae5614dfa7afbfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 17:25:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d4b2ccbcd3e1410d91629d3dd4310ebb : SUCCESS in 2m 29s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/df788993b51f4c229a00b16814c4da8e : SUCCESS in 5m 41s (non-voting)\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/e1ecfe741d814e11b9e2e7d19c1c7f5e : FAILURE in 17m 24s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/204b8484879342558ebe9af6c56a265d : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eab4546fcc464a87ad3af2c1677ae128 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e875a67147b47a89ae34c81c7d53e54 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ff9e43a65644869b3d6a495dce2683b : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71ce224c9f4f4c18a65d3907ea0e7d0a : SUCCESS in 4m 59s\n- vitrage-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/8e17c3eebb424623a7db85331e55799d : SUCCESS in 35m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e37318166a6b89dec6601e609989d3a284d24cf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-12-14 17:51:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f4f3b93123cee06a6f3a70205098a1bc8f593edd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 18:48:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/b243485d1a944e15aa7fb2e2c63a96be : SUCCESS in 4m 37s (non-voting)\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/bc3341ffe3294d1bbb5b4f449ffe2280 : FAILURE in 7m 04s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69fb9a1f4ef04388a41b157289413a57 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d0ef8cf5dc8419fa4ce89866c169d69 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f145bcb49a94ffaba498e3455e26a8d : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f88227085ad4523b9efd021c277a244 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28c07d59ab4e404f82e0994be9cfdcf8 : SUCCESS in 5m 36s\n- vitrage-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/c07819b03150480f8e32045c85e71fd7 : SUCCESS in 48m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"12ccb748fb2c48faa416407f2f32968116c7f9e0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-12-15 05:25:46.000000000","message":"Topic set to Use-Phantomjs","accounts_in_message":[],"_revision_number":2},{"id":"31eb1f756bea630d1c3283d647987a540bc7f447","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-12-15 05:28:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e8f8b00da3965dc3ca4eeff6c954c957141ced3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 06:32:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/a0e5dcee8bc94c11a72b951e3f1a3e37 : SUCCESS in 4m 57s (non-voting)\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/8a2bb421a2b24b55b135fba8fbffbba5 : FAILURE in 5m 49s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe30bf86cc0b453ab00cc01101fe5284 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c0a834f7c1744b68e06532f1ad2bf13 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9263d4fd2ff4c62ba4f3daeae4ff81e : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a439a36026534feeac0620b32b81c68c : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2582930770fb49dab1cbad7ed1dfeda8 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e6bd20021c144cfab979a704a69da36 : SUCCESS in 3m 10s\n- vitrage-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/afc971e5034f490cbe42ba41d0e00b3c : SUCCESS in 57m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5e69cf2a2c4dd2675d0050236dc14451ccdb866f","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-12-15 15:45:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6cff06759efe317c81f2daada5374e1374971e4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 15:51:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"057e784805711ddfaf39ce11c243b61cae88046f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 15:59:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df7e9fd8d18d46c091b25f791a7e66ce : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cb4277125874508b03bf49a87aacc21 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/546397b0096b41e5a5233bf457e48236 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87343a2a683f463eb4128e9adb5a3b93 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfb7ece89f81489986736f99290ed440 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":3},{"id":"f7831082c81e5c58c40548d00e58c9746889d1a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 15:59:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"35ce42ea78a519c905cc40dc61c36ff5c71ded82","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 16:02:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce75e02537564d0c89c3f61a09b62304 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c1451e3a7b7424bb5178be4de9b5410 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ba3fb0c5fa1bce2432808f1af1a9a9b8919a36ae","revisions":{"de339c1f8db2589f00856c54e3d35d0c8c6292b3":{"kind":"REWORK","_number":1,"created":"2020-12-14 16:44:15.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/93/766993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/93/766993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/93/766993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/93/766993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/93/766993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/93/766993/1"}}},"commit":{"parents":[{"commit":"10941239f622536a90a35348f1ebd02283e66bb8","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/10941239f622536a90a35348f1ebd02283e66bb8"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-12-14 16:43:16.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-12-14 16:43:16.000000000","tz":0},"subject":"[DNR] Test npm nodejs job","message":"[DNR] Test npm nodejs job\n\nChange-Id: Id66915db78d788d98645d536fb112dadf9a89aed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/de339c1f8db2589f00856c54e3d35d0c8c6292b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/de339c1f8db2589f00856c54e3d35d0c8c6292b3"}]},"branch":"refs/heads/master"},"b363eea78c5ddd5ffd9f4f6ea2a6878d00518895":{"kind":"REWORK","_number":2,"created":"2020-12-14 17:51:23.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/93/766993/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/93/766993/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/93/766993/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/93/766993/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/93/766993/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/93/766993/2"}}},"commit":{"parents":[{"commit":"10941239f622536a90a35348f1ebd02283e66bb8","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/10941239f622536a90a35348f1ebd02283e66bb8"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-12-14 16:43:16.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-12-14 17:50:21.000000000","tz":0},"subject":"Run npm nodejs job with Phantomjs browser","message":"Run npm nodejs job with Phantomjs browser\n\nThis patch update browser from Chrome to Phantomjs\nin karma.conf.js because nodejs job fails with Chrome\nbrowser as not able to access some URL.\nSo let\u0027s switch to Phantomjs browser now then we can\nuse Chrome browser in future once this issue is resolved.\n\nChange-Id: Id66915db78d788d98645d536fb112dadf9a89aed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/b363eea78c5ddd5ffd9f4f6ea2a6878d00518895"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/b363eea78c5ddd5ffd9f4f6ea2a6878d00518895"}]},"branch":"refs/heads/master"},"ba3fb0c5fa1bce2432808f1af1a9a9b8919a36ae":{"kind":"REWORK","_number":3,"created":"2020-12-15 05:28:32.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/93/766993/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/93/766993/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/93/766993/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/93/766993/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/93/766993/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/93/766993/3"}}},"commit":{"parents":[{"commit":"10941239f622536a90a35348f1ebd02283e66bb8","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/10941239f622536a90a35348f1ebd02283e66bb8"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-12-14 16:43:16.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-12-15 05:27:22.000000000","tz":0},"subject":"Run npm nodejs job with Phantomjs browser","message":"Run npm nodejs job with Phantomjs browser\n\nThis patch update browser from Chrome to Phantomjs\nin karma.conf.js because nodejs job fails with Chrome\nbrowser as not able to access some URL.\nSo let\u0027s switch to Phantomjs browser now then we can\nuse Chrome browser in future once this issue is resolved.\n\nIt also adds libfontconfig1 in bindep.txt as phantomjs\nrequires libfontconfg[1].\n\n[1] https://www.npmjs.com/package/phantomjs-prebuilt#linux-note\n\nChange-Id: Id66915db78d788d98645d536fb112dadf9a89aed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/ba3fb0c5fa1bce2432808f1af1a9a9b8919a36ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/ba3fb0c5fa1bce2432808f1af1a9a9b8919a36ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
