)]}'
{"id":"openstack%2Fvitrage-dashboard~768992","triplet_id":"openstack%2Fvitrage-dashboard~master~Ibaaa569b62bdb7f6f0d1cd4c8510a7d8270bf495","project":"openstack/vitrage-dashboard","branch":"master","topic":"remove-unicode","hashtags":[],"change_id":"Ibaaa569b62bdb7f6f0d1cd4c8510a7d8270bf495","subject":"Remove unicode  from dashboard","status":"MERGED","created":"2021-01-03 07:55:26.000000000","updated":"2021-01-05 07:04:53.000000000","submitted":"2021-01-05 07:04:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768992-remove-unicode","meta_rev_id":"92a12a24c0750a8ff4341c087a2fe2351a065d60","_number":768992,"virtual_id_number":768992,"owner":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-05 07:04:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-05 06:56:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-05 06:56:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-03 08:04:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-05 06:56:24.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"}],"messages":[{"id":"2b0a9d9d2fd58011491c0199c82498ef058e224d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"date":"2021-01-03 07:55:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09f1747a6064ce4a064e98b742316c187bc4ec61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 08:04:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/bed4e49544ac4274b03b3ff92d8a6227 : SUCCESS in 5m 30s (non-voting)\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/34209f446b384f4badf6fc2098a52f09 : FAILURE in 7m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58a64028aa064ef78f6feadcf33d8c73 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64ac609b9851487d988398c75cbb5add : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0db908d4a4d4b70ad37cc26a7fc205b : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"5c8c11206ac70982a98dd827524b4bbe91545e1b","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-01-05 06:56:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"131c73a1a31d89aa1a80b393cd2db7d5c4a3d041","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 06:56:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b544d10641d0aa84fd4ecd4c1d0c2aefd429dd71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 07:04:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d92f633cb7a46b08bfdef412963c617 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b813052f71b042b08fde4b1c13fdff1f : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e55f45f4f614e408d7e566039fc64a9 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"92a12a24c0750a8ff4341c087a2fe2351a065d60","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 07:04:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7603de09b98b93128aea18a3808d2ae1ffe1126b","revisions":{"7603de09b98b93128aea18a3808d2ae1ffe1126b":{"kind":"REWORK","_number":1,"created":"2021-01-03 07:55:26.000000000","uploader":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"ref":"refs/changes/92/768992/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/92/768992/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/92/768992/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/92/768992/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/92/768992/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/92/768992/1"}}},"commit":{"parents":[{"commit":"ba3fb0c5fa1bce2432808f1af1a9a9b8919a36ae","subject":"Run npm nodejs job with Phantomjs browser","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/ba3fb0c5fa1bce2432808f1af1a9a9b8919a36ae"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@aliyun.com","date":"2021-01-03 02:09:41.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@aliyun.com","date":"2021-01-03 07:55:21.000000000","tz":480},"subject":"Remove unicode  from dashboard","message":"Remove unicode  from dashboard\n\nChange-Id: Ibaaa569b62bdb7f6f0d1cd4c8510a7d8270bf495\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/7603de09b98b93128aea18a3808d2ae1ffe1126b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/7603de09b98b93128aea18a3808d2ae1ffe1126b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
