)]}'
{"id":"openstack%2Fvitrage-dashboard~770699","triplet_id":"openstack%2Fvitrage-dashboard~master~Iacc061ecdffe720902caaf1fcd6285ff8e3ef0fc","project":"openstack/vitrage-dashboard","branch":"master","hashtags":[],"change_id":"Iacc061ecdffe720902caaf1fcd6285ff8e3ef0fc","subject":"Revert remove py37","status":"ABANDONED","created":"2021-01-14 02:53:33.000000000","updated":"2021-01-14 06:13:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e217e8bc7d0f0204c8b638c5c8cb36a57196924d","_number":770699,"virtual_id_number":770699,"owner":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-01-14 03:49:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-14 03:49:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"284bac54f949b4c5a9c7631813ac8bd7b8e64003","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-14 02:53:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c978b4dd16b2b4aa160a8798d87ce729faf170ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 03:49:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/05d8ace4f2ad4557a49b47dcfada99a7 : SUCCESS in 1m 53s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/02f28ad6b7cb406da1aa652fc1f5cbe9 : SUCCESS in 5m 46s (non-voting)\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/82f12a07d21944b5807c5b8630698bb3 : FAILURE in 6m 45s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/343c7d8d0bc64a1a81a7422386433200 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83343e650b5243c68e56157ce74aa12e : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a1997f7106e45769ce98c1e7963a63a : SUCCESS in 7m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b23c0ece7494091be3a716a043becc1 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a0ae960ddf64e5b899217f052e8c174 : SUCCESS in 6m 29s\n- vitrage-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/528d282293ff41c482a092bec228dc9a : SUCCESS in 46m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e217e8bc7d0f0204c8b638c5c8cb36a57196924d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-01-14 06:13:22.000000000","message":"Abandoned\n\nhttps://review.opendev.org/c/openstack/vitrage-dashboard/+/770715","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7b4e489e0c28cf188845160312cbbf03699542c5","revisions":{"7b4e489e0c28cf188845160312cbbf03699542c5":{"kind":"REWORK","_number":1,"created":"2021-01-14 02:53:33.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/99/770699/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/99/770699/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/99/770699/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/99/770699/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/99/770699/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/99/770699/1"}}},"commit":{"parents":[{"commit":"9ea52c936223d043fd7a92a443c6d48b55101dab","subject":"Merge \"Update TOX_CONSTRAINTS_FILE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/9ea52c936223d043fd7a92a443c6d48b55101dab"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-14 02:52:23.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2021-01-14 02:53:19.000000000","tz":480},"subject":"Revert remove py37","message":"Revert remove py37\n\nrevert commit 6acaab010eccb4d4cff3a5127b942065ed7cdae8\n\nReason: Python 3.7 *is* supported because it falls\nin the range of supported Python versions (3.6-3.8) [1].\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019648.html\n\nChange-Id: Iacc061ecdffe720902caaf1fcd6285ff8e3ef0fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/7b4e489e0c28cf188845160312cbbf03699542c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/7b4e489e0c28cf188845160312cbbf03699542c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
