)]}'
{"id":"openstack%2Fvitrage-dashboard~801186","triplet_id":"openstack%2Fvitrage-dashboard~master~I6bf98adecb772896c6243a90271a1b258569a4df","project":"openstack/vitrage-dashboard","branch":"master","hashtags":[],"change_id":"I6bf98adecb772896c6243a90271a1b258569a4df","subject":"Migrate from testr to stestr","status":"MERGED","created":"2021-07-17 18:43:49.000000000","updated":"2021-07-18 05:40:06.000000000","submitted":"2021-07-18 05:38:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801186","meta_rev_id":"feb8683860a6d5d435c03683819643aa02c20ee4","_number":801186,"virtual_id_number":801186,"owner":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-18 05:38:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-18 05:30:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-18 05:30:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-17 19:22:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-17 19:29:07.000000000","updated_by":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2021-07-17 19:29:07.000000000","updated_by":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2021-07-17 19:29:07.000000000","updated_by":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"b55da22deca6f93db5815afd781b7c1655d4c3e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"date":"2021-07-17 18:43:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9db28018e57f690ef7fee3083784a2554ed4c63b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-17 19:22:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/e5bc42fae3b64fc2b1dfb0d6ee6328d9 : SUCCESS in 6m 05s (non-voting)\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/3bf19d913df54a36a5111be7ee350f76 : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21e488b65c7142c6be42b93e4feed034 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b941f47860194972976deddf2d84b287 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/038e934756c54a4b8ee5417be0dd54f6 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46f24189f3af4bd08fe5225ddaa5d485 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e5c76884cf34759a9b8773d28ca7cc1 : SUCCESS in 7m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c6c850180cfd4cd382f7600566d69a3e : SUCCESS in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fecfcbc26d5429ca85f7639e8631c18 : SUCCESS in 4m 46s\n- vitrage-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/959992f17b6c43f4854f9e67d241e4d5 : SUCCESS in 37m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6262a1d969906b5d842bed66c057772d50aed7b0","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-07-18 05:30:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5f96f4810afb01d9268937e6f5cade2a3e0881aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-18 05:30:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"88edf1f36a5cfb51a7e34e915e0b934db6f3f2c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-18 05:38:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a659995f3a2e4ad19e54106c9e50531e : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/814767ca353f409a847be91316560513 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07db2078ffad450ebce0a8705313cac2 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eb0d2d0862a4566b13a378dcdf74f8b : SUCCESS in 7m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/772146343bf842688fd86257916d6eeb : SUCCESS in 2m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8171eac4812b45a0ad47ff81b983f0bb : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"68ceccda07015045ce438d056c1522130a2a6f98","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-18 05:38:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"feb8683860a6d5d435c03683819643aa02c20ee4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-18 05:40:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f55562c822854225a059e52ca278d8ad : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a26dea282cb446388b2e4e3f92de89a : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"556a43ef3193a1fa5ea127c64f98c0d5e9972afa","revisions":{"556a43ef3193a1fa5ea127c64f98c0d5e9972afa":{"kind":"REWORK","_number":1,"created":"2021-07-17 18:43:49.000000000","uploader":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"ref":"refs/changes/86/801186/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/86/801186/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/86/801186/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/86/801186/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/86/801186/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/86/801186/1"}}},"commit":{"parents":[{"commit":"0c56491a7eb193353f7ec8d36374a02439ec2205","subject":"Merge \"Drop horizon-nodejs10-jobs template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/0c56491a7eb193353f7ec8d36374a02439ec2205"}]}],"author":{"name":"Anand Bhat","email":"anandgvbhat@gmail.com","date":"2021-07-17 18:35:10.000000000","tz":330},"committer":{"name":"Anand Bhat","email":"anandgvbhat@gmail.com","date":"2021-07-17 18:35:10.000000000","tz":330},"subject":"Migrate from testr to stestr","message":"Migrate from testr to stestr\n\n* Replace .testr.conf by .stestr.conf for migration and update\n  .gitignore,test-requirements files accordingly\n* Use 3.18.0 as the minversion for tox\n\nChange-Id: I6bf98adecb772896c6243a90271a1b258569a4df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/556a43ef3193a1fa5ea127c64f98c0d5e9972afa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/556a43ef3193a1fa5ea127c64f98c0d5e9972afa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
