)]}'
{"id":"openstack%2Fvitrage-dashboard~870922","triplet_id":"openstack%2Fvitrage-dashboard~master~Ic12cddd2fddee64ce115f209b1e10da3b66c689f","project":"openstack/vitrage-dashboard","branch":"master","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2023-01-18 11:32:58.000000000","reason":"Change was submitted"},"19134":{"account":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"last_update":"2023-01-18 11:17:46.000000000","reason":"\u003cGERRIT_ACCOUNT_19134\u003e replied on the change","reason_account":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}}},"hashtags":[],"change_id":"Ic12cddd2fddee64ce115f209b1e10da3b66c689f","subject":"Update tox.ini for tox4","status":"MERGED","created":"2023-01-18 10:27:25.000000000","updated":"2023-01-18 11:34:01.000000000","submitted":"2023-01-18 11:32:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870922","meta_rev_id":"8236e1be0051ed11fb4bb19fd231e635a5425c3e","_number":870922,"virtual_id_number":870922,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-18 11:32:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-18 11:17:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-18 11:17:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-18 11:16:45.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2023-01-18 11:18:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"24b218b7c7653149911cba2ba100af32a395cdae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-01-18 10:27:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6df0a02dfb4ef0d70f342afbd4ab6c24daa14ea0","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2023-01-18 11:17:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"60e0b0144c72925550218bbf51f3445f207ae04b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 11:18:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d16915a6b4946e390298f3bf224b0d3\n\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/4cfcc130770640cabdcde54dc0c5a262 : SUCCESS in 4m 52s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/20199d7f04b743bfa1b3085f1665c019 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/839a46df7a3746c78fa5bbb523d71fdd : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e21a8f3cf2c430a971b2b709aab96fb : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8808281b22594ca29ae561d851a09436 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cfe7f7e7122e4509a0dcbb0a880aa646 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2484e3213af3485a8155165ea053432f : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa4c3fba8b3e4623a9608f11fc6f5905 : SUCCESS in 3m 24s\n- vitrage-dashboard-integration-tests https://zuul.opendev.org/t/openstack/build/fd9e048db3b341f8b80b0c8ba6792af6 : SUCCESS in 32m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d3a91b912c07483dd4e645234dc5760547a825a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 11:18:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2f2bee29114304a91bcce6b244e377131d4c13f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 11:32:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75a80118ed334b3692c9dea7cf0d8b36\n\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/1f983a0de1fd4843b69ad158bfba685f : SUCCESS in 4m 15s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/d9759f1b0ba043bcaf7cdcb828c1f526 : SUCCESS in 8m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2cd888aff4224a08808cd6347635bd6f : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5a1ae5cc5a74a08887dc4616f299636 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05eb3b281c38459591614c6c13078457 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ce9c966f5084a0c91be497ac6721e90 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c044d6f66a14844ade04a99ee618658 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c9473e1687840a29f8206550882fcb3 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"cfaebf41b6c3a96240534576e5a2af23a85e0ac9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 11:32:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8236e1be0051ed11fb4bb19fd231e635a5425c3e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 11:34:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8bb0b95396142fa856a38bedacff00b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abca3287c7c9479d8ce7da44bb7fcc77 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8295ee495634228b5104dc8117d5817 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"935e28fa0e8457d67b027e3e353db6738f0e5ed8","revisions":{"935e28fa0e8457d67b027e3e353db6738f0e5ed8":{"kind":"REWORK","_number":1,"created":"2023-01-18 10:27:25.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/22/870922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/22/870922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/22/870922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/22/870922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/22/870922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/22/870922/1"}}},"commit":{"parents":[{"commit":"2cef4bafdaf34fba3c197b5c8a8a19ddcdc08e60","subject":"Update ``bindep.txt`` to sync with the horizon","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/2cef4bafdaf34fba3c197b5c8a8a19ddcdc08e60"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-01-18 10:26:31.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-01-18 10:26:31.000000000","tz":330},"subject":"Update tox.ini for tox4","message":"Update tox.ini for tox4\n\nOpenStack CI now uses tox4 by default. This patch drop\n\"skipsdist \u003d True\". When skipsdist\u003dTrue is specified, a\ntarget application is not installed into a tox env, but\nthere seems no reason to do so. In various other OpenStack\nprojects, a target applicatin needs to be installed into a\ntox vnev, for example, in docs and/or linter jobs and it\ncauses failures with tox4. So dropping it would avoid potential\nfuture failures.\n\nChange-Id: Ic12cddd2fddee64ce115f209b1e10da3b66c689f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/935e28fa0e8457d67b027e3e353db6738f0e5ed8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/935e28fa0e8457d67b027e3e353db6738f0e5ed8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}}]}],"submit_requirements":[]}
