)]}'
{"id":"openstack%2Fvitrage-dashboard~961124","triplet_id":"openstack%2Fvitrage-dashboard~stable%2F2025.2~Ibc74574560c2aa8d5d5db183dac4e6f234ba4542","project":"openstack/vitrage-dashboard","branch":"stable/2025.2","topic":"create-2025.2","hashtags":[],"change_id":"Ibc74574560c2aa8d5d5db183dac4e6f234ba4542","subject":"Update .gitreview for stable/2025.2","status":"MERGED","created":"2025-09-15 09:36:29.000000000","updated":"2025-09-15 11:11:14.000000000","submitted":"2025-09-15 11:10:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961124-create-2025.2","meta_rev_id":"fd9948525813a05bde9dcb88bc57a5a3757059a8","_number":961124,"virtual_id_number":961124,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-15 11:10:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2025-09-15 10:55:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2025-09-15 10:55:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-15 10:23:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-15 10:55:02.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"e9c5e93be91715d6dbdbe755e900305741c6eae3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2025-09-15 09:36:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"841220f3874d4f6db09ef487abd0b7a8075592d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 10:23:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86af317e13544eacb5ca2c7dd8f6cee0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9494cb52113d4da98ceeef7eab90a8ba : SUCCESS in 6m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0acc069a1fda4b879f81eaa015dd7ec8 : SUCCESS in 4m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f8837c543e04bca8ae2bf66fa277f0b : SUCCESS in 5m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/80f30731bf164c1f9ec27fea4a74f1f5 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d95dd8b90bc4e8980221647465bf351 : SUCCESS in 7m 41s","accounts_in_message":[],"_revision_number":1},{"id":"c9debdabb43caee07929743ea30dde0a01543866","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-15 10:55:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"621d213c8bcaecb4aa9cbd304ebbcf38c0881392","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 10:55:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0429a49da48be37812281494adbccfdb07943a72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 11:10:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/541e57c232fd40308543f8acec06c3e2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83bc94c9293e493cafa907933ed33ee2 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3000bb7c5dc74e56802b6006fbbe13d8 : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6f3dfceedf94b4caec9e08e592f399b : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5319adb21d264e6181c2c1f20ffd4a21 : SUCCESS in 8m 31s","accounts_in_message":[],"_revision_number":1},{"id":"dd8bf469bb6d2325e95f7840dafe20909fb2634a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 11:10:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fd9948525813a05bde9dcb88bc57a5a3757059a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 11:11:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f67bb4bac5b74d6cbd2cda4aea79751b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6e1b4f5be664e02a38e166b4932bc88 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2d587291da862dae2f1e449128c9cd68ca7d813c","revisions":{"2d587291da862dae2f1e449128c9cd68ca7d813c":{"kind":"REWORK","_number":1,"created":"2025-09-15 09:36:29.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/24/961124/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-dashboard","ref":"refs/changes/24/961124/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/24/961124/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/24/961124/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-dashboard refs/changes/24/961124/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-dashboard refs/changes/24/961124/1"}}},"commit":{"parents":[{"commit":"6e823b39eadc1bd93b84f1ca880262344c195324","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/6e823b39eadc1bd93b84f1ca880262344c195324"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-09-15 09:36:28.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-09-15 09:36:28.000000000","tz":0},"subject":"Update .gitreview for stable/2025.2","message":"Update .gitreview for stable/2025.2\n\nChange-Id: Ibc74574560c2aa8d5d5db183dac4e6f234ba4542\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/2d587291da862dae2f1e449128c9cd68ca7d813c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-dashboard/commit/2d587291da862dae2f1e449128c9cd68ca7d813c"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
