)]}'
{"id":"openstack%2Fvitrage-specs~698995","triplet_id":"openstack%2Fvitrage-specs~master~Iac7477a9cfce0458b55d40d5219a315c47abee27","project":"openstack/vitrage-specs","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"Iac7477a9cfce0458b55d40d5219a315c47abee27","subject":"[ussuri][goal] Drop python 2.7 support","status":"MERGED","created":"2019-12-13 18:56:28.000000000","updated":"2019-12-30 12:37:35.000000000","submitted":"2019-12-30 12:35:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698995-1577709358381-d94f54cc","meta_rev_id":"286084e2006e45c154b98f40cfc814cd23a03a38","_number":698995,"virtual_id_number":698995,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-30 12:35:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-30 12:28:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-30 12:28:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-30 12:28:01.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-12-30 12:35:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f8569fe9c722207dd34d96ac216eecb72df1a77","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-12-13 18:56:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6fe46d644599acbb466f1c6d304f61b6b0da03f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 19:01:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddf2977ced764213afde0dadeea81b23 : FAILURE in 3m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ea9351f69782f0d60fbf9051ff0f2f321b2d2d24","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-30 11:49:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c4ed0baba6cf5dc70f2ff62a34e6f47e45c239f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-30 11:53:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9ecf4a869644c60a70be2608c9f6832 : FAILURE in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"15ae53eeae860e4dcb93639cf77aab01da012fe9","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-30 11:59:52.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6976592ebbf1e071cdc780531730f0717be88df8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-30 12:03:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bc334ba4e0c4d3e994b607b6dd57acc : FAILURE in 3m 21s","accounts_in_message":[],"_revision_number":2},{"id":"00c692757498597a49fe5fd7e08a5382674e0883","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-30 12:19:02.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3eb451218c317728ac74d373ca61b1cf8f478f97","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-30 12:21:42.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"0c49b36311de58e082c897c58d03ebe3a11a0c3c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-30 12:22:13.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"280c7ff8457e1466a550a23ba59a5cbf4caa95e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-30 12:27:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6654471adf44528bed46f51cf5f3f16 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":5},{"id":"79f39c58a590d07e79350bc9d69823263b5deb79","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-30 12:28:01.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"2bd4a0de7e8a78f48b40c30e501bc8d2d5dd2173","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-30 12:28:12.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7a3dc180ccc4871ccaedc54128f9c1a3992bd63b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-30 12:35:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"e58e02668e700de7aeb5a15681f543aab5fe1615","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-30 12:35:58.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ade096d84a734a1f80a5f95694e2d0d9 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":5},{"id":"0e423e90417cda99f58c45bea713832be0759283","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-30 12:37:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/d4de568a4a9c4f1fb97d4d0b46134ed0 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c7b937010d2cf13e5cf0cfab80fcca8d9e4e74e1","revisions":{"5d9bf3c3d1e4bb131dbcc6b7ac7a177cfca98a78":{"kind":"REWORK","_number":1,"created":"2019-12-13 18:56:28.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/95/698995/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-specs","ref":"refs/changes/95/698995/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-specs refs/changes/95/698995/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-specs refs/changes/95/698995/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-specs refs/changes/95/698995/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-specs refs/changes/95/698995/1"}}},"commit":{"parents":[{"commit":"c1c43bb1c44740767f8232185ba3df41f7875905","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-specs/commit/c1c43bb1c44740767f8232185ba3df41f7875905"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-12-13 18:50:29.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-12-13 18:56:27.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support","message":"[ussuri][goal] Drop python 2.7 support\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nspecs repo either has py27 job or requirement or tox env.\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: Iac7477a9cfce0458b55d40d5219a315c47abee27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-specs/commit/5d9bf3c3d1e4bb131dbcc6b7ac7a177cfca98a78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-specs/commit/5d9bf3c3d1e4bb131dbcc6b7ac7a177cfca98a78"}]},"branch":"refs/heads/master"},"f8b5a81f7b909ec73dae956d9ca1a05ca289a544":{"kind":"REWORK","_number":2,"created":"2019-12-30 11:59:52.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/95/698995/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-specs","ref":"refs/changes/95/698995/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-specs refs/changes/95/698995/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-specs refs/changes/95/698995/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-specs refs/changes/95/698995/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-specs refs/changes/95/698995/2"}}},"commit":{"parents":[{"commit":"c1c43bb1c44740767f8232185ba3df41f7875905","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-specs/commit/c1c43bb1c44740767f8232185ba3df41f7875905"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-12-13 18:50:29.000000000","tz":0},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-30 11:59:46.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support","message":"[ussuri][goal] Drop python 2.7 support\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nspecs repo either has py27 job or requirement or tox env.\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: Iac7477a9cfce0458b55d40d5219a315c47abee27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-specs/commit/f8b5a81f7b909ec73dae956d9ca1a05ca289a544"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-specs/commit/f8b5a81f7b909ec73dae956d9ca1a05ca289a544"}]},"branch":"refs/heads/master"},"6183acf4a266ab54f5e8d9a4b2e17c32e58ceb22":{"kind":"REWORK","_number":3,"created":"2019-12-30 12:19:02.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/95/698995/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-specs","ref":"refs/changes/95/698995/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-specs refs/changes/95/698995/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-specs refs/changes/95/698995/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-specs refs/changes/95/698995/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-specs refs/changes/95/698995/3"}}},"commit":{"parents":[{"commit":"c1c43bb1c44740767f8232185ba3df41f7875905","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-specs/commit/c1c43bb1c44740767f8232185ba3df41f7875905"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-12-13 18:50:29.000000000","tz":0},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-30 12:18:57.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support","message":"[ussuri][goal] Drop python 2.7 support\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nspecs repo either has py27 job or requirement or tox env.\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: Iac7477a9cfce0458b55d40d5219a315c47abee27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-specs/commit/6183acf4a266ab54f5e8d9a4b2e17c32e58ceb22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-specs/commit/6183acf4a266ab54f5e8d9a4b2e17c32e58ceb22"}]},"branch":"refs/heads/master"},"6b8753acf8d4ceb1f8710dbd96d85fd75c12cc04":{"kind":"REWORK","_number":4,"created":"2019-12-30 12:21:42.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/95/698995/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-specs","ref":"refs/changes/95/698995/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-specs refs/changes/95/698995/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-specs refs/changes/95/698995/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-specs refs/changes/95/698995/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-specs refs/changes/95/698995/4"}}},"commit":{"parents":[{"commit":"c1c43bb1c44740767f8232185ba3df41f7875905","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-specs/commit/c1c43bb1c44740767f8232185ba3df41f7875905"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-12-13 18:50:29.000000000","tz":0},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-30 12:21:39.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support","message":"[ussuri][goal] Drop python 2.7 support\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nspecs repo either has py27 job or requirement or tox env.\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: Iac7477a9cfce0458b55d40d5219a315c47abee27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-specs/commit/6b8753acf8d4ceb1f8710dbd96d85fd75c12cc04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-specs/commit/6b8753acf8d4ceb1f8710dbd96d85fd75c12cc04"}]},"branch":"refs/heads/master"},"c7b937010d2cf13e5cf0cfab80fcca8d9e4e74e1":{"kind":"REWORK","_number":5,"created":"2019-12-30 12:22:13.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/95/698995/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-specs","ref":"refs/changes/95/698995/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-specs refs/changes/95/698995/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-specs refs/changes/95/698995/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-specs refs/changes/95/698995/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-specs refs/changes/95/698995/5"}}},"commit":{"parents":[{"commit":"c1c43bb1c44740767f8232185ba3df41f7875905","subject":"Replace git.openstack.org URLs with opendev.org URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-specs/commit/c1c43bb1c44740767f8232185ba3df41f7875905"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-12-13 18:50:29.000000000","tz":0},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-30 12:22:09.000000000","tz":0},"subject":"[ussuri][goal] Drop python 2.7 support","message":"[ussuri][goal] Drop python 2.7 support\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nspecs repo either has py27 job or requirement or tox env.\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: Iac7477a9cfce0458b55d40d5219a315c47abee27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-specs/commit/c7b937010d2cf13e5cf0cfab80fcca8d9e4e74e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-specs/commit/c7b937010d2cf13e5cf0cfab80fcca8d9e4e74e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
