)]}'
{"id":"openstack%2Fvitrage-specs~768902","triplet_id":"openstack%2Fvitrage-specs~master~If0ce15d90e698ca58a6f00ffb6675250f375cca8","project":"openstack/vitrage-specs","branch":"master","topic":"remove-unicode","hashtags":[],"change_id":"If0ce15d90e698ca58a6f00ffb6675250f375cca8","subject":"Remove unicode  from specs","status":"MERGED","created":"2021-01-03 01:59:13.000000000","updated":"2021-01-05 07:09:27.000000000","submitted":"2021-01-05 07:08:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768902-remove-unicode","meta_rev_id":"96d8acb3d439ef7cb13aab94442ecfde41afbe79","_number":768902,"virtual_id_number":768902,"owner":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-05 07:08:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-05 06:58:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-05 06:58:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-03 02:06:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-05 06:58:49.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"}],"messages":[{"id":"9f6839243dae27c6dd0b74424b4d79e740b352d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"date":"2021-01-03 01:59:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ddc7cdb47bb3ed65729b88017d4284535e22b661","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 02:06:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d204d7fd53a842539a9b4d203b7dcab7 : FAILURE in 3m 50s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/mitaka/aodh-notifier.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/mitaka/aodh-notifier.rst","accounts_in_message":[],"_revision_number":1},{"id":"6bafa5c5bb81647481daf5aa7c925b932413b0bd","author":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"date":"2021-01-03 08:05:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"85e19d76a42ed2987cefcebc4ce402b8457aa37d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 08:13:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90631687c0964b8e8c2fa72a4c178b8f : FAILURE in 3m 19s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/mitaka/aodh-notifier.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/mitaka/aodh-notifier.rst","accounts_in_message":[],"_revision_number":1},{"id":"82fdafa81e44f8cde6eb23f00b62810fe4702196","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-01-05 06:54:42.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e8a51178d3bc0b9e765b33e1dcdb7c67e12ef548","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-01-05 06:58:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"aba2d2705a5bded1ade3aff3670ed5c21622d67a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 07:01:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b1cf6e592e74291a235310fc093ae29 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":2},{"id":"9eb62028f434eecc5f14fc6e6a44b7da5c4ff7fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 07:01:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"376c85a18e799111659e11440009ee7bac0422f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 07:08:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd3f29993076477cbff9e02be1fb588f : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":2},{"id":"60b129357cdfe294bb8c0a2f57840bffbe15b23c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 07:08:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"96d8acb3d439ef7cb13aab94442ecfde41afbe79","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 07:09:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/0f1543f5c7e84cc98d10314e600db6d0 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"099e9fd543897da1607aff4bd1fa69a4c2e04942","revisions":{"10819db522c7a06549c0925e201d949089d601ac":{"kind":"REWORK","_number":1,"created":"2021-01-03 01:59:13.000000000","uploader":{"_account_id":32577,"name":"Hongtao Qi","display_name":"Qitao","email":"qihongtao@aliyun.com","username":"qihongtao03"},"ref":"refs/changes/02/768902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-specs","ref":"refs/changes/02/768902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-specs refs/changes/02/768902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-specs refs/changes/02/768902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-specs refs/changes/02/768902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-specs refs/changes/02/768902/1"}}},"commit":{"parents":[{"commit":"ce370bb2d66e8f0503cb234741ccf3dd71ff83c0","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-specs/commit/ce370bb2d66e8f0503cb234741ccf3dd71ff83c0"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@aliyun.com","date":"2021-01-03 01:56:01.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@aliyun.com","date":"2021-01-03 01:56:01.000000000","tz":480},"subject":"Remove unicode  from specs","message":"Remove unicode  from specs\n\nChange-Id: If0ce15d90e698ca58a6f00ffb6675250f375cca8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-specs/commit/10819db522c7a06549c0925e201d949089d601ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-specs/commit/10819db522c7a06549c0925e201d949089d601ac"}]},"branch":"refs/heads/master"},"099e9fd543897da1607aff4bd1fa69a4c2e04942":{"kind":"REWORK","_number":2,"created":"2021-01-05 06:54:42.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/02/768902/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-specs","ref":"refs/changes/02/768902/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-specs refs/changes/02/768902/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-specs refs/changes/02/768902/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-specs refs/changes/02/768902/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-specs refs/changes/02/768902/2"}}},"commit":{"parents":[{"commit":"ce370bb2d66e8f0503cb234741ccf3dd71ff83c0","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-specs/commit/ce370bb2d66e8f0503cb234741ccf3dd71ff83c0"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@aliyun.com","date":"2021-01-03 01:56:01.000000000","tz":480},"committer":{"name":"Eyal","email":"eyalb1@gmail.com","date":"2021-01-05 06:54:32.000000000","tz":0},"subject":"Remove unicode  from specs","message":"Remove unicode  from specs\n\nChange-Id: If0ce15d90e698ca58a6f00ffb6675250f375cca8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-specs/commit/099e9fd543897da1607aff4bd1fa69a4c2e04942"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-specs/commit/099e9fd543897da1607aff4bd1fa69a4c2e04942"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
