)]}'
{"id":"openstack%2Fvitrage-tempest-plugin~539135","triplet_id":"openstack%2Fvitrage-tempest-plugin~master~Icfc60a26a7b2328c8f425c2b70844de661e4073d","project":"openstack/vitrage-tempest-plugin","branch":"master","topic":"eyalb/dvsm","hashtags":[],"change_id":"Icfc60a26a7b2328c8f425c2b70844de661e4073d","subject":"disable dsvm jobs","status":"MERGED","created":"2018-01-30 08:24:35.000000000","updated":"2018-01-30 09:34:24.000000000","submitted":"2018-01-30 09:34:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"539135-1517304864340-b4c38317","meta_rev_id":"ef431e92af1f400a6b7ce2fe5350d89cd9b5c6d3","_number":539135,"virtual_id_number":539135,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2018-01-30 09:34:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2018-01-30 08:57:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2018-01-30 08:57:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-01-30 09:34:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c409276e0e90e60ff104f89f4a7301b3e7e3f779","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2018-01-30 08:24:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d62ea428dea4c59d349b1a3e7cbb3deb0e76d98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-30 08:31:32.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/vitrage-tempest-plugin on branch master.  The error\nwas:\n\n  Configuration file is not a list.  Each zuul.yaml configuration\n  file must be a list of items, for example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  Ensure that every item starts with \"- \" so that it is parsed as a\n  YAML list.","accounts_in_message":[],"_revision_number":1},{"id":"f34e35ffb7e7f351f80d00c92817837ca9badac4","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2018-01-30 08:33:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c357cca8ecc8a7e6981d9447ca05e0d8295a6bbe","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2018-01-30 08:57:36.000000000","message":"Patch Set 2: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a40be4155b60fd2c00bf838ae31cce8a44b85ee7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-30 09:12:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/35/539135/2/check/openstack-tox-pep8/36d7a16/ : SUCCESS in 9m 18s","accounts_in_message":[],"_revision_number":2},{"id":"f1c67721bf6d593a5ef11def663ed55659766bb6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-30 09:12:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dbfda07c77b18c20e9d9d4f8c54053c222feb238","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-30 09:34:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d721305cda975dae4d0fd9d6d5e078ed1d24a7ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-30 09:34:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/35/539135/2/gate/openstack-tox-pep8/e4022d2/ : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1d00f20f4dbbe78e3aaf15ceddd50cd112fa1c04","revisions":{"a1589d6c025b9c49ed39f808556663b1c3fdc0ab":{"kind":"REWORK","_number":1,"created":"2018-01-30 08:24:35.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/35/539135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-tempest-plugin","ref":"refs/changes/35/539135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-tempest-plugin refs/changes/35/539135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-tempest-plugin refs/changes/35/539135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-tempest-plugin refs/changes/35/539135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-tempest-plugin refs/changes/35/539135/1"}}},"commit":{"parents":[{"commit":"ba58ed46b9a196617f05f1209b36f50d48e87ac0","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-tempest-plugin/commit/ba58ed46b9a196617f05f1209b36f50d48e87ac0"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2018-01-30 08:24:15.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2018-01-30 08:24:15.000000000","tz":120},"subject":"disable dsvm jobs","message":"disable dsvm jobs\n\nuntil we figure out how to resolve the circular dependency\n\nChange-Id: Icfc60a26a7b2328c8f425c2b70844de661e4073d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-tempest-plugin/commit/a1589d6c025b9c49ed39f808556663b1c3fdc0ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-tempest-plugin/commit/a1589d6c025b9c49ed39f808556663b1c3fdc0ab"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ba58ed46b9a196617f05f1209b36f50d48e87ac0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"1d00f20f4dbbe78e3aaf15ceddd50cd112fa1c04":{"kind":"REWORK","_number":2,"created":"2018-01-30 08:33:20.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/35/539135/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-tempest-plugin","ref":"refs/changes/35/539135/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-tempest-plugin refs/changes/35/539135/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-tempest-plugin refs/changes/35/539135/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-tempest-plugin refs/changes/35/539135/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-tempest-plugin refs/changes/35/539135/2"}}},"commit":{"parents":[{"commit":"ba58ed46b9a196617f05f1209b36f50d48e87ac0","subject":"Merge \"Updated from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-tempest-plugin/commit/ba58ed46b9a196617f05f1209b36f50d48e87ac0"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2018-01-30 08:24:15.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2018-01-30 08:33:06.000000000","tz":120},"subject":"disable dsvm jobs","message":"disable dsvm jobs\n\nuntil we figure out how to resolve the circular dependency\n\nChange-Id: Icfc60a26a7b2328c8f425c2b70844de661e4073d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-tempest-plugin/commit/1d00f20f4dbbe78e3aaf15ceddd50cd112fa1c04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-tempest-plugin/commit/1d00f20f4dbbe78e3aaf15ceddd50cd112fa1c04"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ba58ed46b9a196617f05f1209b36f50d48e87ac0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
