)]}'
{"id":"openstack%2Fvitrage-tempest-plugin~935419","triplet_id":"openstack%2Fvitrage-tempest-plugin~master~I4686c922aaa639ae9dce33e1fb0fac9da78b6a93","project":"openstack/vitrage-tempest-plugin","branch":"master","topic":"migrate-to-noble","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-11-29 07:49:21.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-11-29 08:41:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4686c922aaa639ae9dce33e1fb0fac9da78b6a93","subject":"Update gate jobs as per the 2025.1 cycle testing runtime","status":"MERGED","created":"2024-11-16 04:19:36.000000000","updated":"2024-11-29 08:44:32.000000000","submitted":"2024-11-29 08:41:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935419-migrate-to-noble","meta_rev_id":"5a0268f617c4069ce9c7e9b78ccdf7d6f2743cbe","_number":935419,"virtual_id_number":935419,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-29 08:41:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2024-11-29 07:49:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2024-11-29 07:49:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-16 05:34:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-27 09:05:28.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"92bfaa712f37ef001042ccf775f35f0d6d55baf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-16 04:19:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afeccbe3d30cd7edf972f565c98b935dc8669317","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 05:34:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33d3261268db4069b48c192b0853b30e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03c7226a4520432788936b4a1abdc6c1 : SUCCESS in 2m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/749fa37c5a5e429b964fc95709d5fb30 : FAILURE in 2m 26s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/90ae283746e242d3b1b8fe8e8d79b5d0 : FAILURE in 17m 46s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/a4c24061a19d4df993568e673e23564d : SUCCESS in 1h 00m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7ba1614fa868d87f2bea8e952c0cc3684354910a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-27 04:19:47.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"38f8b948f6c2354b7719a445e9b7abfc23b7ee2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 05:07:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d9c6bb4f51f4e58b85291f8ce923299\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8352121d74c54b479ba619d2f2b8693d : SUCCESS in 2m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ae64691309646218de24320a10e4361 : SUCCESS in 2m 31s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/3568332ab9284bf2ae4eb883464de1fa : SUCCESS in 37m 35s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/4218432f64024923835c83383888f7c8 : SUCCESS in 45m 32s","accounts_in_message":[],"_revision_number":2},{"id":"3965493acf06bfa7b090af798412b2d754a8ab51","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-27 09:05:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"380fa4fc30682200113bff7a2de30e49c216f196","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 09:05:41.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4369ea2f73064c709ba130aa5c941e8e\n\n\nWarning:\n  Change 932156 in project openstack/tempest does not share a change queue with 935419 in project openstack/vitrage-tempest-plugin","accounts_in_message":[],"_revision_number":2},{"id":"08f65a3e00a86d09b7c089d5fe6df19be23cea79","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-28 22:55:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9efcb5b62262c68f4aff1d995044016dedf2ce6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-28 22:56:15.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"876da558b43e2730f2f2b0a433536958a0b37618","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 23:43:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d6edf7aa3e848699ed09cd8800d3795\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97d6c34473774b188bdab15a02f73206 : SUCCESS in 1m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5564638b48304c4cb799ac4b6293ce83 : SUCCESS in 1m 56s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/820c26fd21e14b4d8e0344ed13a88846 : SUCCESS in 34m 12s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/1549c240b7d7453f8a7387fadb9e9a05 : FAILURE in 45m 47s","accounts_in_message":[],"_revision_number":3},{"id":"720ab1dfeaa33027d56c67b8f9041137074085cf","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-29 05:25:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"325b508b5bdc35ebc63c8d07a8cf80cb9305eac2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 06:22:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e2975e7253f43d1afb386074f358d6d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f1f11c5f8fd401a9f212ac5ddb169e6 : SUCCESS in 1m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/859d7afa78db451cb800d96e2a228bf2 : SUCCESS in 2m 03s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/cf59251d733147e0a7c8ddaf5bfa34d2 : SUCCESS in 50m 59s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/68dbd0d9a8dd450c818bcef7a1ec8ab5 : SUCCESS in 41m 41s","accounts_in_message":[],"_revision_number":3},{"id":"12cbee0b073de8efa49d8d0973ef3690f291ee23","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-11-29 07:49:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"428a75f6a9e011143d2a3a882b5fd3ae11053a99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 07:50:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"292f7a46c03665e8920b9989693994f5b651f26e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 08:41:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44901d32d78f49a6a32c581611c8bac3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18de2c92c0434f1bbe6ad5b7c053a9df : SUCCESS in 2m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/914d2a0c2a6f4f31b742012aeb051267 : SUCCESS in 2m 38s\n- vitrage-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/22b1bfcb32564e1c848e14290f0232ce : SUCCESS in 42m 59s\n- vitrage-tempest-plugin-datasources https://zuul.opendev.org/t/openstack/build/686718788f09494fb1bcfc730777486a : SUCCESS in 50m 32s","accounts_in_message":[],"_revision_number":3},{"id":"b584d8438276766b9a1f877619396c18418a51e6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 08:41:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5a0268f617c4069ce9c7e9b78ccdf7d6f2743cbe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 08:44:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d3e21b230404a31a7b3eabbde9ba64f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de8a2517ccfe47148b735693f734788b : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"20cfc6ed0265b41c310aa08bf20503b0ef4df069","revisions":{"c86cdee35801a701ef0519b9db8819fb75773692":{"kind":"REWORK","_number":1,"created":"2024-11-16 04:19:36.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/19/935419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-tempest-plugin","ref":"refs/changes/19/935419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-tempest-plugin refs/changes/19/935419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-tempest-plugin refs/changes/19/935419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-tempest-plugin refs/changes/19/935419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-tempest-plugin refs/changes/19/935419/1"}}},"commit":{"parents":[{"commit":"053ec3f6fce8134d11b67b9e4e8d55fce89e249b","subject":"Drop unnecessary \u0027x\u0027 bit from doc config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-tempest-plugin/commit/053ec3f6fce8134d11b67b9e4e8d55fce89e249b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-16 04:17:05.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-16 04:19:34.000000000","tz":-480},"subject":"Update gate jobs as per the 2025.1 cycle testing runtime","message":"Update gate jobs as per the 2025.1 cycle testing runtime\n\nAs per 2025.1 testing runtime[1], we need to test on Ubuntu\nNoble (which will be taken care by depends-on tempest and devstack\npatches to move base jobs to Noble) and at least single job to run on\nUbutnu Jammy (for smooth upgrade from previous releases).\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/932156\nTracking: https://etherpad.opendev.org/p/migrate-to-noble\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: I4686c922aaa639ae9dce33e1fb0fac9da78b6a93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-tempest-plugin/commit/c86cdee35801a701ef0519b9db8819fb75773692"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-tempest-plugin/commit/c86cdee35801a701ef0519b9db8819fb75773692"}]},"branch":"refs/heads/master"},"a572a839059806792ea01b263acbcbaa8db23919":{"kind":"REWORK","_number":2,"created":"2024-11-27 04:19:47.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/19/935419/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-tempest-plugin","ref":"refs/changes/19/935419/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-tempest-plugin refs/changes/19/935419/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-tempest-plugin refs/changes/19/935419/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-tempest-plugin refs/changes/19/935419/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-tempest-plugin refs/changes/19/935419/2"}}},"commit":{"parents":[{"commit":"053ec3f6fce8134d11b67b9e4e8d55fce89e249b","subject":"Drop unnecessary \u0027x\u0027 bit from doc config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-tempest-plugin/commit/053ec3f6fce8134d11b67b9e4e8d55fce89e249b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-16 04:17:05.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-11-27 04:19:46.000000000","tz":0},"subject":"Update gate jobs as per the 2025.1 cycle testing runtime","message":"Update gate jobs as per the 2025.1 cycle testing runtime\n\nAs per 2025.1 testing runtime[1], we need to test on Ubuntu\nNoble (which will be taken care by depends-on tempest and devstack\npatches to move base jobs to Noble) and at least single job to run on\nUbutnu Jammy (for smooth upgrade from previous releases).\n\nBecause of bug#2088358, we need to enabling the devstack\u0027s GLOBAL_ENV.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/932156\nTracking: https://etherpad.opendev.org/p/migrate-to-noble\n\nCloses-Bug: #2088358\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n\nChange-Id: I4686c922aaa639ae9dce33e1fb0fac9da78b6a93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-tempest-plugin/commit/a572a839059806792ea01b263acbcbaa8db23919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-tempest-plugin/commit/a572a839059806792ea01b263acbcbaa8db23919"}]},"branch":"refs/heads/master"},"20cfc6ed0265b41c310aa08bf20503b0ef4df069":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-11-28 22:56:15.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/19/935419/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage-tempest-plugin","ref":"refs/changes/19/935419/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage-tempest-plugin refs/changes/19/935419/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage-tempest-plugin refs/changes/19/935419/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage-tempest-plugin refs/changes/19/935419/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage-tempest-plugin refs/changes/19/935419/3"}}},"commit":{"parents":[{"commit":"053ec3f6fce8134d11b67b9e4e8d55fce89e249b","subject":"Drop unnecessary \u0027x\u0027 bit from doc config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-tempest-plugin/commit/053ec3f6fce8134d11b67b9e4e8d55fce89e249b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-16 04:17:05.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-11-28 22:56:15.000000000","tz":0},"subject":"Update gate jobs as per the 2025.1 cycle testing runtime","message":"Update gate jobs as per the 2025.1 cycle testing runtime\n\nAs per 2025.1 testing runtime[1], we need to test on Ubuntu\nNoble (which will be taken care by tempest and devstack\npatches[2] to move base jobs to Noble) and at least single job to run on\nUbutnu Jammy (for smooth upgrade from previous releases).\n\nBecause of bug#2088358, we need to enabling the devstack\u0027s GLOBAL_ENV.\n\nTracking: https://etherpad.opendev.org/p/migrate-to-noble\n\nCloses-Bug: #2088358\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.1.html\n[2] https://review.opendev.org/c/openstack/tempest/+/932156\n\nChange-Id: I4686c922aaa639ae9dce33e1fb0fac9da78b6a93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-tempest-plugin/commit/20cfc6ed0265b41c310aa08bf20503b0ef4df069"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage-tempest-plugin/commit/20cfc6ed0265b41c310aa08bf20503b0ef4df069"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
