)]}'
{"id":"openstack%2Fvitrage~293036","triplet_id":"openstack%2Fvitrage~master~I3207b095c230d70a6e0967dbb4573585e74fb05e","project":"openstack/vitrage","branch":"master","topic":"load-plugins-from-conf-file","hashtags":[],"change_id":"I3207b095c230d70a6e0967dbb4573585e74fb05e","subject":"load-plugins-from-conf-file-update","status":"ABANDONED","created":"2016-03-15 16:41:46.000000000","updated":"2016-03-20 11:17:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bd28d31f0d7133dfe4458534b0baf77ca3c3ee91","_number":293036,"virtual_id_number":293036,"owner":{"_account_id":20667,"name":"Noam Bloom","email":"noam00bloom@gmail.com","username":"nbloom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":1,"date":"2016-03-15 16:52:57.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-03-15 16:52:57.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"2b2e956a0449f7c7d028499476454b6b275435ec","author":{"_account_id":20667,"name":"Noam Bloom","email":"noam00bloom@gmail.com","username":"nbloom"},"date":"2016-03-15 16:41:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5fa48420f0d4e8f80a162c5f531d65ac809e0e7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-15 16:52:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-vitrage-pep8 http://logs.openstack.org/36/293036/1/check/gate-vitrage-pep8/878c384/ : SUCCESS in 4m 42s\n- gate-vitrage-docs http://docs-draft.openstack.org/36/293036/1/check/gate-vitrage-docs/563b652//doc/build/html/ : SUCCESS in 2m 25s\n- gate-vitrage-python27 http://logs.openstack.org/36/293036/1/check/gate-vitrage-python27/039f4b9/ : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":1},{"id":"ac9f1656c5a85b1fcfb32ecc17306789f3986303","author":{"_account_id":20667,"name":"Noam Bloom","email":"noam00bloom@gmail.com","username":"nbloom"},"date":"2016-03-20 11:17:09.000000000","message":"Abandoned\n\nOld commit","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8a28d1293a04a11aee2d187ff1ccc7b4b018460","revisions":{"a8a28d1293a04a11aee2d187ff1ccc7b4b018460":{"kind":"REWORK","_number":1,"created":"2016-03-15 16:41:46.000000000","uploader":{"_account_id":20667,"name":"Noam Bloom","email":"noam00bloom@gmail.com","username":"nbloom"},"ref":"refs/changes/36/293036/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/36/293036/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/36/293036/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/36/293036/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/36/293036/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/36/293036/1"}}},"commit":{"parents":[{"commit":"2411f6c693aae77ab2e622a862da08ee12c0b5f3","subject":"Merge \"refactor _find_rca_index\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/2411f6c693aae77ab2e622a862da08ee12c0b5f3"}]}],"author":{"name":"Noam Bloom","email":"noam.bloom.ext@nokia.com","date":"2016-03-15 16:40:05.000000000","tz":0},"committer":{"name":"Noam Bloom","email":"noam.bloom.ext@nokia.com","date":"2016-03-15 16:41:44.000000000","tz":0},"subject":"load-plugins-from-conf-file-update","message":"load-plugins-from-conf-file-update\n\nChange-Id: I3207b095c230d70a6e0967dbb4573585e74fb05e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/a8a28d1293a04a11aee2d187ff1ccc7b4b018460"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/a8a28d1293a04a11aee2d187ff1ccc7b4b018460"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2411f6c693aae77ab2e622a862da08ee12c0b5f3","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
