)]}'
{"id":"openstack%2Fvitrage~445156","triplet_id":"openstack%2Fvitrage~master~I786c24761f554d9681b530af65b02ed53809bdaf","project":"openstack/vitrage","branch":"master","topic":"openstack/requirements","hashtags":[],"change_id":"I786c24761f554d9681b530af65b02ed53809bdaf","subject":"Updated from global requirements","status":"MERGED","created":"2017-03-13 19:44:05.000000000","updated":"2017-03-14 09:41:28.000000000","submitted":"2017-03-14 09:41:28.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4e3a4dd7eb89e7a038e7f0a5a3a013eb9a0e8ec7","_number":445156,"virtual_id_number":445156,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},{"value":2,"date":"2017-03-14 09:41:28.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"recommended":{"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},"all":[{"value":1,"date":"2017-03-14 08:54:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-03-14 08:19:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":1,"date":"2017-03-14 08:53:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-03-14 08:19:37.000000000","updated_by":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"reviewer":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"state":"REVIEWER"},{"updated":"2017-03-14 08:53:11.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2017-03-14 08:54:17.000000000","updated_by":{"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},"reviewer":{"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},"state":"REVIEWER"},{"updated":"2017-03-14 09:41:28.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"026ed797dc6482d010f57025b48405efb1b55844","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-03-13 19:44:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4729e8117a5e178ce7d938aa63d20b2e3614ad4a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-13 20:47:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-vitrage-docs-ubuntu-xenial http://docs-draft.openstack.org/56/445156/1/check/gate-vitrage-docs-ubuntu-xenial/fd50a99//doc/build/html/ : SUCCESS in 9m 25s\n- gate-vitrage-pep8-ubuntu-xenial http://logs.openstack.org/56/445156/1/check/gate-vitrage-pep8-ubuntu-xenial/c2327e1/ : SUCCESS in 8m 54s\n- gate-vitrage-python27-ubuntu-xenial http://logs.openstack.org/56/445156/1/check/gate-vitrage-python27-ubuntu-xenial/e244c36/ : SUCCESS in 10m 03s\n- gate-vitrage-python35 http://logs.openstack.org/56/445156/1/check/gate-vitrage-python35/9aebede/ : SUCCESS in 8m 53s\n- gate-vitrage-requirements http://logs.openstack.org/56/445156/1/check/gate-vitrage-requirements/5628f47/ : SUCCESS in 8m 50s\n- gate-vitrage-releasenotes http://docs-draft.openstack.org/56/445156/1/check/gate-vitrage-releasenotes/9d3b562//releasenotes/build/html/ : SUCCESS in 8m 07s\n- gate-vitrage-dsvm-api-ubuntu-xenial http://logs.openstack.org/56/445156/1/check/gate-vitrage-dsvm-api-ubuntu-xenial/624eb1c/ : SUCCESS in 46m 19s\n- gate-vitrage-dsvm-datasources-ubuntu-xenial http://logs.openstack.org/56/445156/1/check/gate-vitrage-dsvm-datasources-ubuntu-xenial/036a183/ : SUCCESS in 44m 21s","accounts_in_message":[],"_revision_number":1},{"id":"91a05c1d2ffea4adbb11728e918901678adb5ca6","author":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"date":"2017-03-14 08:19:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4360ac972d45727bc409ef5533980732d8cd0d71","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2017-03-14 08:53:11.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"db9ca36a4229af9d062de3f87a6e4dc19dfbbb60","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-14 08:53:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"afd5f9c7ca1545c9ea853701c12188c8738bf888","author":{"_account_id":23825,"name":"chenyingnan","display_name":"chenyingnan","email":"chenyingnan01@inspur.com","username":"Cady_Chen"},"date":"2017-03-14 08:54:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2cb5eab3ce503694c8fd58425ede4a41ae2de938","date":"2017-03-14 09:41:28.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"fff6637912c0b2ce777c620c416fab81270d3cb3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-14 09:41:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-vitrage-docs-ubuntu-xenial http://docs-draft.openstack.org/56/445156/1/gate/gate-vitrage-docs-ubuntu-xenial/736b331//doc/build/html/ : SUCCESS in 5m 09s\n- gate-vitrage-pep8-ubuntu-xenial http://logs.openstack.org/56/445156/1/gate/gate-vitrage-pep8-ubuntu-xenial/4ef32b0/ : SUCCESS in 6m 09s\n- gate-vitrage-python27-ubuntu-xenial http://logs.openstack.org/56/445156/1/gate/gate-vitrage-python27-ubuntu-xenial/a80e77e/ : SUCCESS in 4m 42s\n- gate-vitrage-python35 http://logs.openstack.org/56/445156/1/gate/gate-vitrage-python35/e0dcbf3/ : SUCCESS in 5m 18s\n- gate-vitrage-requirements http://logs.openstack.org/56/445156/1/gate/gate-vitrage-requirements/83fc50e/ : SUCCESS in 2m 46s\n- gate-vitrage-releasenotes http://docs-draft.openstack.org/56/445156/1/gate/gate-vitrage-releasenotes/a61f629//releasenotes/build/html/ : SUCCESS in 3m 25s\n- gate-vitrage-dsvm-api-ubuntu-xenial http://logs.openstack.org/56/445156/1/gate/gate-vitrage-dsvm-api-ubuntu-xenial/e65eea5/ : SUCCESS in 47m 35s\n- gate-vitrage-dsvm-datasources-ubuntu-xenial http://logs.openstack.org/56/445156/1/gate/gate-vitrage-dsvm-datasources-ubuntu-xenial/c85a3f6/ : SUCCESS in 46m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"60cdf650334fb219a07e6388a03d0bdb3293129f","revisions":{"60cdf650334fb219a07e6388a03d0bdb3293129f":{"kind":"REWORK","_number":1,"created":"2017-03-13 19:44:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/56/445156/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/56/445156/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/56/445156/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/56/445156/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/56/445156/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/56/445156/1"}}},"commit":{"parents":[{"commit":"a330ef92ed862c2b4e556c4f6a6f9fc4bccb4097","subject":"deatils is a dictionary don\u0027t do loads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/a330ef92ed862c2b4e556c4f6a6f9fc4bccb4097"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-03-13 19:44:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-03-13 19:44:02.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I786c24761f554d9681b530af65b02ed53809bdaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/60cdf650334fb219a07e6388a03d0bdb3293129f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/60cdf650334fb219a07e6388a03d0bdb3293129f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a330ef92ed862c2b4e556c4f6a6f9fc4bccb4097","is_merged_in_target_branch":true,"change_id":"I6f9d7de1c34362090f56ec03c8fe03d5d06fa1fb","change_number":444879,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
