)]}'
{"id":"openstack%2Fvitrage~484334","triplet_id":"openstack%2Fvitrage~master~I4962812206ea170c855a2574083b7b136d067850","project":"openstack/vitrage","branch":"master","topic":"assertEqual","hashtags":[],"change_id":"I4962812206ea170c855a2574083b7b136d067850","subject":"Fixed order of arguments in assertEqual","status":"MERGED","created":"2017-07-17 11:33:41.000000000","updated":"2017-07-19 12:22:16.000000000","submitted":"2017-07-19 12:22:16.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"df8be0d87af30ff54d9ffc5520d2fa2b66964006","_number":484334,"virtual_id_number":484334,"owner":{"_account_id":25006,"name":"sudhir agarwal","email":"sudhir.agarwal@nectechnologies.in","username":"sudhir_agarwal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2017-07-19 12:22:16.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":21414,"name":"Yujun Zhang","email":"zhang.yujunz@zte.com.cn","username":"yujunz"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21414,"name":"Yujun Zhang","email":"zhang.yujunz@zte.com.cn","username":"yujunz"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-07-19 07:23:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21414,"name":"Yujun Zhang","email":"zhang.yujunz@zte.com.cn","username":"yujunz"},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":21414,"name":"Yujun Zhang","email":"zhang.yujunz@zte.com.cn","username":"yujunz"},{"value":1,"date":"2017-07-19 11:27:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21414,"name":"Yujun Zhang","email":"zhang.yujunz@zte.com.cn","username":"yujunz"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-07-19 07:23:22.000000000","updated_by":{"_account_id":21414,"name":"Yujun Zhang","email":"zhang.yujunz@zte.com.cn","username":"yujunz"},"reviewer":{"_account_id":21414,"name":"Yujun Zhang","email":"zhang.yujunz@zte.com.cn","username":"yujunz"},"state":"REVIEWER"},{"updated":"2017-07-19 11:27:09.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2017-07-19 12:22:16.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"09f23bb79e870f35975421f72468526e675b8646","author":{"_account_id":25006,"name":"sudhir agarwal","email":"sudhir.agarwal@nectechnologies.in","username":"sudhir_agarwal"},"date":"2017-07-17 11:33:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd7e9b8d34d3ab51a47fb47a2cd2d1aaa3a90e19","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-17 12:21:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-vitrage-docs-ubuntu-xenial http://docs-draft.openstack.org/34/484334/1/check/gate-vitrage-docs-ubuntu-xenial/b336e1c//doc/build/html/ : SUCCESS in 2m 25s\n- gate-vitrage-pep8-ubuntu-xenial http://logs.openstack.org/34/484334/1/check/gate-vitrage-pep8-ubuntu-xenial/26470a7/ : SUCCESS in 4m 33s\n- gate-vitrage-python27-ubuntu-xenial http://logs.openstack.org/34/484334/1/check/gate-vitrage-python27-ubuntu-xenial/ca75168/ : FAILURE in 3m 07s\n- gate-vitrage-python35 http://logs.openstack.org/34/484334/1/check/gate-vitrage-python35/3a2c0b9/ : FAILURE in 4m 30s\n- gate-vitrage-dsvm-api-ubuntu-xenial http://logs.openstack.org/34/484334/1/check/gate-vitrage-dsvm-api-ubuntu-xenial/e181c3d/ : SUCCESS in 30m 49s\n- gate-vitrage-dsvm-datasources-ubuntu-xenial http://logs.openstack.org/34/484334/1/check/gate-vitrage-dsvm-datasources-ubuntu-xenial/15df8f5/ : SUCCESS in 46m 52s","accounts_in_message":[],"_revision_number":1},{"id":"cf6046272d6dd960484e355bc7065f4af8c475bf","author":{"_account_id":25006,"name":"sudhir agarwal","email":"sudhir.agarwal@nectechnologies.in","username":"sudhir_agarwal"},"date":"2017-07-17 12:58:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"432475c3ba13bfb25000527da7ce1c8fec245ba5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-17 13:48:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-vitrage-docs-ubuntu-xenial http://docs-draft.openstack.org/34/484334/2/check/gate-vitrage-docs-ubuntu-xenial/04c98fe//doc/build/html/ : SUCCESS in 5m 00s\n- gate-vitrage-pep8-ubuntu-xenial http://logs.openstack.org/34/484334/2/check/gate-vitrage-pep8-ubuntu-xenial/9e83c10/ : SUCCESS in 3m 55s\n- gate-vitrage-python27-ubuntu-xenial http://logs.openstack.org/34/484334/2/check/gate-vitrage-python27-ubuntu-xenial/8e3354e/ : SUCCESS in 3m 59s\n- gate-vitrage-python35 http://logs.openstack.org/34/484334/2/check/gate-vitrage-python35/3aa9c28/ : SUCCESS in 2m 54s\n- gate-vitrage-dsvm-api-ubuntu-xenial http://logs.openstack.org/34/484334/2/check/gate-vitrage-dsvm-api-ubuntu-xenial/fd05877/ : SUCCESS in 47m 46s\n- gate-vitrage-dsvm-datasources-ubuntu-xenial http://logs.openstack.org/34/484334/2/check/gate-vitrage-dsvm-datasources-ubuntu-xenial/3fc606f/ : SUCCESS in 45m 21s","accounts_in_message":[],"_revision_number":2},{"id":"20c9cae1cfa49f2df0933711bac0aa9f1e7ac90e","author":{"_account_id":21414,"name":"Yujun Zhang","email":"zhang.yujunz@zte.com.cn","username":"yujunz"},"date":"2017-07-19 01:03:10.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nIt seems not all tests are corrected.","accounts_in_message":[],"_revision_number":2},{"id":"4c51e768eeb05feab75ff664c5a1d637a6660bf0","author":{"_account_id":25006,"name":"sudhir agarwal","email":"sudhir.agarwal@nectechnologies.in","username":"sudhir_agarwal"},"date":"2017-07-19 06:18:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5d0e99fad0c080e2dcf9d4255aa2f16919672953","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-19 07:04:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-vitrage-docs-ubuntu-xenial http://docs-draft.openstack.org/34/484334/3/check/gate-vitrage-docs-ubuntu-xenial/ca23d97//doc/build/html/ : SUCCESS in 4m 13s\n- gate-vitrage-pep8-ubuntu-xenial http://logs.openstack.org/34/484334/3/check/gate-vitrage-pep8-ubuntu-xenial/8eee0e7/ : SUCCESS in 5m 23s\n- gate-vitrage-python27-ubuntu-xenial http://logs.openstack.org/34/484334/3/check/gate-vitrage-python27-ubuntu-xenial/2f4d3e5/ : SUCCESS in 3m 26s\n- gate-vitrage-python35 http://logs.openstack.org/34/484334/3/check/gate-vitrage-python35/e1d5b9f/ : SUCCESS in 4m 21s\n- gate-vitrage-dsvm-api-ubuntu-xenial http://logs.openstack.org/34/484334/3/check/gate-vitrage-dsvm-api-ubuntu-xenial/d4a322a/ : SUCCESS in 44m 03s\n- gate-vitrage-dsvm-datasources-ubuntu-xenial http://logs.openstack.org/34/484334/3/check/gate-vitrage-dsvm-datasources-ubuntu-xenial/4d21baf/ : SUCCESS in 33m 10s","accounts_in_message":[],"_revision_number":3},{"id":"5dfad7eacf4be5c58fdd322ceac34d2659d2b673","author":{"_account_id":21414,"name":"Yujun Zhang","email":"zhang.yujunz@zte.com.cn","username":"yujunz"},"date":"2017-07-19 07:23:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"139710e7add62ad38f29de7e41918b095e1f49f7","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2017-07-19 11:27:09.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"082a65ab6f7fbb06a9b11cc88bcc1eb91f273bf8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-19 11:27:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"86e2ec6e5a99c649f6dc0aaf705d28db9f155821","date":"2017-07-19 12:22:16.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":3},{"id":"674859c02333168377243e3325984b64c3199298","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-19 12:22:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-vitrage-docs-ubuntu-xenial http://docs-draft.openstack.org/34/484334/3/gate/gate-vitrage-docs-ubuntu-xenial/107f549//doc/build/html/ : SUCCESS in 4m 42s\n- gate-vitrage-pep8-ubuntu-xenial http://logs.openstack.org/34/484334/3/gate/gate-vitrage-pep8-ubuntu-xenial/2bc08c9/ : SUCCESS in 3m 36s\n- gate-vitrage-python27-ubuntu-xenial http://logs.openstack.org/34/484334/3/gate/gate-vitrage-python27-ubuntu-xenial/f036e1e/ : SUCCESS in 3m 02s\n- gate-vitrage-python35 http://logs.openstack.org/34/484334/3/gate/gate-vitrage-python35/3d67343/ : SUCCESS in 2m 44s\n- gate-vitrage-dsvm-api-ubuntu-xenial http://logs.openstack.org/34/484334/3/gate/gate-vitrage-dsvm-api-ubuntu-xenial/a442de9/ : SUCCESS in 33m 10s\n- gate-vitrage-dsvm-datasources-ubuntu-xenial http://logs.openstack.org/34/484334/3/gate/gate-vitrage-dsvm-datasources-ubuntu-xenial/1ed2d50/ : SUCCESS in 53m 34s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6491f4c078fe73dc301074e33527a4e12419a041","revisions":{"60e72b5d5d518bcebc2b9e3727d2800e43c4f4fb":{"kind":"REWORK","_number":1,"created":"2017-07-17 11:33:41.000000000","uploader":{"_account_id":25006,"name":"sudhir agarwal","email":"sudhir.agarwal@nectechnologies.in","username":"sudhir_agarwal"},"ref":"refs/changes/34/484334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/34/484334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/34/484334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/34/484334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/34/484334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/34/484334/1"}}},"commit":{"parents":[{"commit":"e405beae70cada9328bf7b2d7d0012253e600e8c","subject":"Use openstackdocstheme","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/e405beae70cada9328bf7b2d7d0012253e600e8c"}]}],"author":{"name":"sudhir_agarwal","email":"sudhir.agarwal@nectechnologies.in","date":"2017-07-17 11:27:01.000000000","tz":330},"committer":{"name":"sudhir_agarwal","email":"sudhir.agarwal@nectechnologies.in","date":"2017-07-17 11:28:12.000000000","tz":330},"subject":"Fixed order of arguments in assertEqual","message":"Fixed order of arguments in assertEqual\n\nSome tests used incorrect order of arguments in\nassertEqual(observed, expected). The correct order expected\nby testtool is assertEqual(expected, observed).\n\nChange-Id: I4962812206ea170c855a2574083b7b136d067850\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/60e72b5d5d518bcebc2b9e3727d2800e43c4f4fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/60e72b5d5d518bcebc2b9e3727d2800e43c4f4fb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e405beae70cada9328bf7b2d7d0012253e600e8c","is_merged_in_target_branch":true,"change_id":"I6e206edb28a741284ab3978c05819657c49215ef","change_number":484100,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"21f687d8d9d88069e139c4da1086aa47fea4b2f2":{"kind":"REWORK","_number":2,"created":"2017-07-17 12:58:31.000000000","uploader":{"_account_id":25006,"name":"sudhir agarwal","email":"sudhir.agarwal@nectechnologies.in","username":"sudhir_agarwal"},"ref":"refs/changes/34/484334/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/34/484334/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/34/484334/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/34/484334/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/34/484334/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/34/484334/2"}}},"commit":{"parents":[{"commit":"e405beae70cada9328bf7b2d7d0012253e600e8c","subject":"Use openstackdocstheme","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/e405beae70cada9328bf7b2d7d0012253e600e8c"}]}],"author":{"name":"sudhir_agarwal","email":"sudhir.agarwal@nectechnologies.in","date":"2017-07-17 11:27:01.000000000","tz":330},"committer":{"name":"sudhir_agarwal","email":"sudhir.agarwal@nectechnologies.in","date":"2017-07-17 12:52:50.000000000","tz":330},"subject":"Fixed order of arguments in assertEqual","message":"Fixed order of arguments in assertEqual\n\nSome tests used incorrect order of arguments in\nassertEqual(observed, expected). The correct order expected\nby testtool is assertEqual(expected, observed).\n\nChange-Id: I4962812206ea170c855a2574083b7b136d067850\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/21f687d8d9d88069e139c4da1086aa47fea4b2f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/21f687d8d9d88069e139c4da1086aa47fea4b2f2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e405beae70cada9328bf7b2d7d0012253e600e8c","is_merged_in_target_branch":true,"change_id":"I6e206edb28a741284ab3978c05819657c49215ef","change_number":484100,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"6491f4c078fe73dc301074e33527a4e12419a041":{"kind":"REWORK","_number":3,"created":"2017-07-19 06:18:45.000000000","uploader":{"_account_id":25006,"name":"sudhir agarwal","email":"sudhir.agarwal@nectechnologies.in","username":"sudhir_agarwal"},"ref":"refs/changes/34/484334/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/34/484334/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/34/484334/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/34/484334/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/34/484334/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/34/484334/3"}}},"commit":{"parents":[{"commit":"e405beae70cada9328bf7b2d7d0012253e600e8c","subject":"Use openstackdocstheme","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/e405beae70cada9328bf7b2d7d0012253e600e8c"}]}],"author":{"name":"sudhir_agarwal","email":"sudhir.agarwal@nectechnologies.in","date":"2017-07-17 11:27:01.000000000","tz":330},"committer":{"name":"sudhir_agarwal","email":"sudhir.agarwal@nectechnologies.in","date":"2017-07-19 06:21:10.000000000","tz":330},"subject":"Fixed order of arguments in assertEqual","message":"Fixed order of arguments in assertEqual\n\nSome tests used incorrect order of arguments in\nassertEqual(observed, expected). The correct order expected\nby testtool is assertEqual(expected, observed).\n\nChange-Id: I4962812206ea170c855a2574083b7b136d067850\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/6491f4c078fe73dc301074e33527a4e12419a041"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/6491f4c078fe73dc301074e33527a4e12419a041"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e405beae70cada9328bf7b2d7d0012253e600e8c","is_merged_in_target_branch":true,"change_id":"I6e206edb28a741284ab3978c05819657c49215ef","change_number":484100,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
