)]}'
{"id":"openstack%2Fvitrage~536760","triplet_id":"openstack%2Fvitrage~master~If9dd2376382ff7b175932da33c1b85224d03b2d8","project":"openstack/vitrage","branch":"master","topic":"idanh/stop_all_processes","hashtags":[],"change_id":"If9dd2376382ff7b175932da33c1b85224d03b2d8","subject":"bugfix - subprocesses remain alive after vitrage-graph restart","status":"MERGED","created":"2018-01-23 10:54:14.000000000","updated":"2018-01-23 13:46:10.000000000","submitted":"2018-01-23 13:46:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"536760-1516715170125-a327bdb8","meta_rev_id":"4bf687b5a73041ab173a4ccaab37689c63a0e9f8","_number":536760,"virtual_id_number":536760,"owner":{"_account_id":19184,"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","username":"idan-hefetz"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":2,"date":"2018-01-23 13:46:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"all":[{"value":2,"date":"2018-01-23 10:55:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2018-01-23 12:02:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-01-23 10:55:24.000000000","updated_by":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"reviewer":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"state":"REVIEWER"},{"updated":"2018-01-23 12:02:32.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2018-01-23 13:46:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dfc179784153f58ffce5fed1db2098a0f7e43e36","author":{"_account_id":19184,"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","username":"idan-hefetz"},"date":"2018-01-23 10:54:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a97c108fd3fd0346a89ec9501d4c735d63803ec","author":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"date":"2018-01-23 10:55:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ab4e83acaa6d246ae08e8f22938401b8c012252f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-23 12:02:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/60/536760/1/check/openstack-tox-pep8/a1f74be/ : SUCCESS in 5m 13s\n- openstack-tox-py27 http://logs.openstack.org/60/536760/1/check/openstack-tox-py27/9d0541c/ : SUCCESS in 6m 01s\n- openstack-tox-py35 http://logs.openstack.org/60/536760/1/check/openstack-tox-py35/20e46a6/ : SUCCESS in 5m 26s\n- build-openstack-sphinx-docs http://logs.openstack.org/60/536760/1/check/build-openstack-sphinx-docs/2c7fc99/html/ : SUCCESS in 4m 51s\n- vitrage-dsvm-api-py27 http://logs.openstack.org/60/536760/1/check/vitrage-dsvm-api-py27/443c057/ : SUCCESS in 48m 32s\n- vitrage-dsvm-datasources-py27 http://logs.openstack.org/60/536760/1/check/vitrage-dsvm-datasources-py27/0fa9601/ : SUCCESS in 53m 20s\n- vitrage-dsvm-api-py35 http://logs.openstack.org/60/536760/1/check/vitrage-dsvm-api-py35/9c9eb9e/ : SUCCESS in 53m 13s\n- vitrage-dsvm-datasources-py35 http://logs.openstack.org/60/536760/1/check/vitrage-dsvm-datasources-py35/f5116be/ : SUCCESS in 56m 27s","accounts_in_message":[],"_revision_number":1},{"id":"17b0af6cbdee4df1154a522fc663bf171f51c14f","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2018-01-23 12:02:32.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a0c767bd60dc3218aeebaeeffd9aa34106bd89bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-23 12:02:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"047ebaa4c6a379ac1c67f5ef06201f37822bce15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-23 13:46:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/60/536760/1/gate/openstack-tox-pep8/8d785cb/ : SUCCESS in 4m 29s\n- openstack-tox-py27 http://logs.openstack.org/60/536760/1/gate/openstack-tox-py27/a8e8545/ : SUCCESS in 5m 16s\n- openstack-tox-py35 http://logs.openstack.org/60/536760/1/gate/openstack-tox-py35/087a52b/ : SUCCESS in 6m 01s\n- build-openstack-sphinx-docs http://logs.openstack.org/60/536760/1/gate/build-openstack-sphinx-docs/689b9a8/html/ : SUCCESS in 4m 53s\n- vitrage-dsvm-api-py27 http://logs.openstack.org/60/536760/1/gate/vitrage-dsvm-api-py27/452f0ac/ : SUCCESS in 1h 04m 03s\n- vitrage-dsvm-datasources-py27 http://logs.openstack.org/60/536760/1/gate/vitrage-dsvm-datasources-py27/d4b0fbd/ : SUCCESS in 51m 35s\n- vitrage-dsvm-api-py35 http://logs.openstack.org/60/536760/1/gate/vitrage-dsvm-api-py35/5407ab1/ : SUCCESS in 56m 20s\n- vitrage-dsvm-datasources-py35 http://logs.openstack.org/60/536760/1/gate/vitrage-dsvm-datasources-py35/fb3f0ad/ : SUCCESS in 52m 34s","accounts_in_message":[],"_revision_number":1},{"id":"effddf1c2a3d4e251d8a7ab07c74aaa9069e0956","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-23 13:46:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"33a4905cc5a1cdf8df77c33eb529cc785f099f14","revisions":{"33a4905cc5a1cdf8df77c33eb529cc785f099f14":{"kind":"REWORK","_number":1,"created":"2018-01-23 10:54:14.000000000","uploader":{"_account_id":19184,"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","username":"idan-hefetz"},"ref":"refs/changes/60/536760/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/60/536760/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/60/536760/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/60/536760/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/60/536760/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/60/536760/1"}}},"commit":{"parents":[{"commit":"27f3d40465d8b6a1375974994d9bf108f8b92118","subject":"Merge \"Webhook release notes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/27f3d40465d8b6a1375974994d9bf108f8b92118"}]}],"author":{"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","date":"2018-01-23 10:53:50.000000000","tz":0},"committer":{"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","date":"2018-01-23 10:53:50.000000000","tz":0},"subject":"bugfix - subprocesses remain alive after vitrage-graph restart","message":"bugfix - subprocesses remain alive after vitrage-graph restart\n\nChange-Id: If9dd2376382ff7b175932da33c1b85224d03b2d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/33a4905cc5a1cdf8df77c33eb529cc785f099f14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/33a4905cc5a1cdf8df77c33eb529cc785f099f14"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"27f3d40465d8b6a1375974994d9bf108f8b92118","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
