)]}'
{"id":"openstack%2Fvitrage~642256","triplet_id":"openstack%2Fvitrage~master~Ieefa3fd9993e273cebf22a10c238f75e47394322","project":"openstack/vitrage","branch":"master","topic":"eyalb/bionic","hashtags":[],"change_id":"Ieefa3fd9993e273cebf22a10c238f75e47394322","subject":"DNM: Testing vitrage legacy jobs on bionic","status":"ABANDONED","created":"2019-03-10 08:24:45.000000000","updated":"2019-03-18 08:42:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bd264d3d16c0e983133ef12bfa93e0599294c25f","_number":642256,"virtual_id_number":642256,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-03-10 10:07:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-03-10 15:16:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-10 10:07:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"20baff4c4d4fb6d0804b5404f247a75cfa42d5aa","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-03-10 08:24:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7e8616af121ae99864049dbfaeb87a578877e97","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-03-10 08:49:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"acb36b6d6c28b729cd9f7f555160f88878aaf748","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-10 10:07:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/56/642256/2/check/openstack-tox-lower-constraints/de3fc79/ : SUCCESS in 8m 30s\n- openstack-tox-pep8 http://logs.openstack.org/56/642256/2/check/openstack-tox-pep8/089ab3b/ : SUCCESS in 6m 38s\n- openstack-tox-py27 http://logs.openstack.org/56/642256/2/check/openstack-tox-py27/dbdbfb9/ : SUCCESS in 5m 32s\n- openstack-tox-py35 http://logs.openstack.org/56/642256/2/check/openstack-tox-py35/f29110b/ : SUCCESS in 6m 06s\n- openstack-tox-py36 http://logs.openstack.org/56/642256/2/check/openstack-tox-py36/89dcb09/ : SUCCESS in 7m 49s\n- openstack-tox-py37 http://logs.openstack.org/56/642256/2/check/openstack-tox-py37/d3d4805/ : SUCCESS in 6m 19s\n- openstack-tox-docs http://logs.openstack.org/56/642256/2/check/openstack-tox-docs/c29da93/html/ : SUCCESS in 5m 04s\n- vitrage-dsvm-api-py27 http://logs.openstack.org/56/642256/2/check/vitrage-dsvm-api-py27/8b5eddc/ : SUCCESS in 52m 49s\n- vitrage-dsvm-datasources-py27 http://logs.openstack.org/56/642256/2/check/vitrage-dsvm-datasources-py27/af8772a/ : SUCCESS in 1h 05m 08s\n- vitrage-dsvm-mock-py27 http://logs.openstack.org/56/642256/2/check/vitrage-dsvm-mock-py27/6f38190/ : SUCCESS in 50m 56s (non-voting)\n- vitrage-dsvm-api-py35 http://logs.openstack.org/56/642256/2/check/vitrage-dsvm-api-py35/b8aadf3/ : SUCCESS in 59m 03s\n- vitrage-dsvm-datasources-py35 http://logs.openstack.org/56/642256/2/check/vitrage-dsvm-datasources-py35/951f2f6/ : SUCCESS in 1h 01m 15s\n- vitrage-dsvm-mock-py35 http://logs.openstack.org/56/642256/2/check/vitrage-dsvm-mock-py35/1ef85be/ : SUCCESS in 52m 41s (non-voting)\n- grenade-vitrage http://logs.openstack.org/56/642256/2/check/grenade-vitrage/739f7cc/ : SUCCESS in 42m 06s","accounts_in_message":[],"_revision_number":2},{"id":"ed5d98975c950c4d56183bccaef2dfaba3be51cb","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-03-10 15:16:42.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"826647ff4c68e8e00a48653a6f7901084aae1c94","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-03-18 08:42:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"642a2c18765ac218cf5423b088a8de8b5835377e","revisions":{"fbf268ebf98f4c4571ed852ada83e764920524de":{"kind":"REWORK","_number":1,"created":"2019-03-10 08:24:45.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/56/642256/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/56/642256/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/56/642256/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/56/642256/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/56/642256/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/56/642256/1"}}},"commit":{"parents":[{"commit":"2be799b33e4a964394c5cade72f790a3b8b85fd7","subject":"Document the template parameters in version 3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/2be799b33e4a964394c5cade72f790a3b8b85fd7"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-03-10 08:24:32.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-03-10 08:24:32.000000000","tz":120},"subject":"DNM: Testing vitrage legacy jobs on bionic","message":"DNM: Testing vitrage legacy jobs on bionic\n\nChange-Id: Ieefa3fd9993e273cebf22a10c238f75e47394322\nDepends-On: https://review.openstack.org/#/c/641886/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/fbf268ebf98f4c4571ed852ada83e764920524de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/fbf268ebf98f4c4571ed852ada83e764920524de"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2be799b33e4a964394c5cade72f790a3b8b85fd7","is_merged_in_target_branch":true,"change_id":"I5c16efd64ab5893d2aa8d1802d7402ac7d5d1c75","change_number":640701,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"642a2c18765ac218cf5423b088a8de8b5835377e":{"kind":"REWORK","_number":2,"created":"2019-03-10 08:49:53.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/56/642256/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/56/642256/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/56/642256/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/56/642256/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/56/642256/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/56/642256/2"}}},"commit":{"parents":[{"commit":"2be799b33e4a964394c5cade72f790a3b8b85fd7","subject":"Document the template parameters in version 3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/2be799b33e4a964394c5cade72f790a3b8b85fd7"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-03-10 08:24:32.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-03-10 08:49:27.000000000","tz":120},"subject":"DNM: Testing vitrage legacy jobs on bionic","message":"DNM: Testing vitrage legacy jobs on bionic\n\nChange-Id: Ieefa3fd9993e273cebf22a10c238f75e47394322\nDepends-On: https://review.openstack.org/#/c/641886/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/642a2c18765ac218cf5423b088a8de8b5835377e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/642a2c18765ac218cf5423b088a8de8b5835377e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2be799b33e4a964394c5cade72f790a3b8b85fd7","is_merged_in_target_branch":true,"change_id":"I5c16efd64ab5893d2aa8d1802d7402ac7d5d1c75","change_number":640701,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
