)]}'
{"id":"openstack%2Fvitrage~693854","triplet_id":"openstack%2Fvitrage~master~I4add1c99466409171e238d09585a58a4bc9af596","project":"openstack/vitrage","branch":"master","topic":"networkx","hashtags":[],"change_id":"I4add1c99466409171e238d09585a58a4bc9af596","subject":"networkx node attribute is deprecated","status":"MERGED","created":"2019-11-12 14:52:24.000000000","updated":"2019-11-13 10:36:20.000000000","submitted":"2019-11-13 10:34:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693854-1573641298997-60743900","meta_rev_id":"fe83df06bd3ba652cf7393f8be358cdbf9fad35e","_number":693854,"virtual_id_number":693854,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19184,"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","username":"idan-hefetz"},{"value":0,"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-13 10:34:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19184,"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","username":"idan-hefetz"},"all":[{"value":2,"date":"2019-11-13 09:27:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19184,"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","username":"idan-hefetz"},{"value":2,"date":"2019-11-13 08:51:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19184,"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","username":"idan-hefetz"},"all":[{"value":1,"date":"2019-11-13 09:27:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19184,"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","username":"idan-hefetz"},{"value":0,"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"_account_id":19184,"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","username":"idan-hefetz"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-13 08:51:26.000000000","updated_by":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"reviewer":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"state":"REVIEWER"},{"updated":"2019-11-13 09:27:51.000000000","updated_by":{"_account_id":19184,"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","username":"idan-hefetz"},"reviewer":{"_account_id":19184,"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","username":"idan-hefetz"},"state":"REVIEWER"},{"updated":"2019-11-13 10:34:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"24e8117a450c3567baf52476f061adb133d51200","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-12 14:52:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc46db108e1aec5d6c8fca9dd6004645e6529035","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 15:56:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df0d2daee41446fea021f16bde0820cc : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93b9beff954c4435b55d889778eb3d77 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29ec2ee400c14168af77e0fada5e5624 : SUCCESS in 7m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81ec2cf347e044ec92cf352c1520b47a : POST_FAILURE in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a41ffab1bf1454896364ac2c048e823 : SUCCESS in 4m 52s\n- vitrage-dsvm-api-py3 https://zuul.opendev.org/t/openstack/build/b7d8d881439f4389ae5ede52ebae98ee : SUCCESS in 55m 46s\n- vitrage-dsvm-datasources-py3 https://zuul.opendev.org/t/openstack/build/1f97f17ace714345948a22de7d5ca417 : SUCCESS in 1h 02m 54s\n- vitrage-dsvm-mock-py3 https://zuul.opendev.org/t/openstack/build/399801d1542e4e9291d340c145be2166 : SUCCESS in 50m 11s (non-voting)\n- grenade-vitrage https://zuul.opendev.org/t/openstack/build/bb1c5d1d8dc24407a33043f7e40d8744 : SUCCESS in 45m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e66812a6aa24d41d07da013d5787623394a1c703","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-12 16:35:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"202e2d0f48c22679ecff5b194851b8aff2149946","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 18:04:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc6231497dbc4561914b5f4e7b97cbe5 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a64139104824358beecdcc470ce8b99 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e31187bc0d544659d548dd45bc36412 : SUCCESS in 6m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/480875037b6d497397240c3da87fbc86 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caea469384f84d8faa417432e7fe625b : SUCCESS in 4m 34s\n- vitrage-dsvm-api-py3 https://zuul.opendev.org/t/openstack/build/ad09abde32454064b1622bca0b554aa5 : SUCCESS in 49m 19s\n- vitrage-dsvm-datasources-py3 https://zuul.opendev.org/t/openstack/build/b04ca247a93344999620237b7212d2fe : SUCCESS in 1h 06m 34s\n- vitrage-dsvm-mock-py3 https://zuul.opendev.org/t/openstack/build/6abe98efd85a4e119bcad36000cf321e : SUCCESS in 1h 06m 53s (non-voting)\n- grenade-vitrage https://zuul.opendev.org/t/openstack/build/1f85bde554d34bd3a4a7883c58e8de01 : SUCCESS in 59m 52s","accounts_in_message":[],"_revision_number":2},{"id":"0ad291c2b2c7495ed98cd52b17243b25fa83d695","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-13 08:35:01.000000000","message":"Topic set to networkx","accounts_in_message":[],"_revision_number":2},{"id":"4dea477294acbd9e3344d0bf2e521b767a1807d9","author":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"date":"2019-11-13 08:51:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"988d3ec1685d73a6edeb73c67587bdfb0b89263a","author":{"_account_id":19184,"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","username":"idan-hefetz"},"date":"2019-11-13 09:27:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"73d665d789df17fe16fd2f3b7619bf128f05e4b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 09:28:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8fee5a2f559c5d99c92d22cefb6d605add3ade17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 10:34:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9bcac32459c409086f4f41a4962d159 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aedc02ab29c94bf99f881abd4874349b : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/618f5022da5f44ca8284601b59b06c3b : SUCCESS in 5m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d590ed32936e4dedabc1f48f448e0391 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7221f40acd4a4621a76d686f1e31c49f : SUCCESS in 5m 29s\n- vitrage-dsvm-api-py3 https://zuul.opendev.org/t/openstack/build/3885e2f2d74f4c2ba4fe8fb0de228746 : SUCCESS in 52m 15s\n- vitrage-dsvm-datasources-py3 https://zuul.opendev.org/t/openstack/build/d06657d75a30403ab1d61ba44323bfb4 : SUCCESS in 1h 05m 28s\n- grenade-vitrage https://zuul.opendev.org/t/openstack/build/e27dcf6cfec44e089ef8d3fbfbf3a0a1 : SUCCESS in 48m 55s","accounts_in_message":[],"_revision_number":2},{"id":"428e3bd3b05c2d650ab8104055981147deaf5b7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 10:34:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fe83df06bd3ba652cf7393f8be358cdbf9fad35e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 10:36:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54db1d726bcd482caae8fc85d95192d7 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b2e03404aa722eb029ec1867a97d1a38c257c558","revisions":{"c4300f8f06288c3c2b2a9cb858902bdc4ea53889":{"kind":"REWORK","_number":1,"created":"2019-11-12 14:52:24.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/54/693854/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/54/693854/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/693854/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/693854/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/693854/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/54/693854/1"}}},"commit":{"parents":[{"commit":"1aac4646f8158dcd0e1ddb3359fa4ac9e31ee6aa","subject":"Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/1aac4646f8158dcd0e1ddb3359fa4ac9e31ee6aa"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-12 14:52:11.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-12 14:52:11.000000000","tz":120},"subject":"networkx node attribute is deprecated","message":"networkx node attribute is deprecated\n\n\u0027node\u0027 was changed to \u0027nodes\u0027\n\nChange-Id: I4add1c99466409171e238d09585a58a4bc9af596\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/c4300f8f06288c3c2b2a9cb858902bdc4ea53889"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/c4300f8f06288c3c2b2a9cb858902bdc4ea53889"}]},"branch":"refs/heads/master"},"b2e03404aa722eb029ec1867a97d1a38c257c558":{"kind":"REWORK","_number":2,"created":"2019-11-12 16:35:55.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/54/693854/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/54/693854/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/693854/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/693854/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/54/693854/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/54/693854/2"}}},"commit":{"parents":[{"commit":"1aac4646f8158dcd0e1ddb3359fa4ac9e31ee6aa","subject":"Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/1aac4646f8158dcd0e1ddb3359fa4ac9e31ee6aa"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-12 16:34:49.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-11-12 16:35:20.000000000","tz":120},"subject":"networkx node attribute is deprecated","message":"networkx node attribute is deprecated\n\n\u0027node\u0027 was changed to \u0027nodes\u0027\n\nChange-Id: I4add1c99466409171e238d09585a58a4bc9af596\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/b2e03404aa722eb029ec1867a97d1a38c257c558"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/b2e03404aa722eb029ec1867a97d1a38c257c558"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
