)]}'
{"id":"openstack%2Fvitrage~694212","triplet_id":"openstack%2Fvitrage~master~I67e4c59fc1170b07e8f7c50416de8e8598152f09","project":"openstack/vitrage","branch":"master","topic":"bugfix_update_neighbor","hashtags":[],"change_id":"I67e4c59fc1170b07e8f7c50416de8e8598152f09","subject":"Bugfix - Processor check if neighbors are newer before updating them","status":"MERGED","created":"2019-11-14 08:39:04.000000000","updated":"2019-11-18 11:12:06.000000000","submitted":"2019-11-18 11:10:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694212-1574075453476-bdb6f180","meta_rev_id":"2ecb5dff4797435b0b21497e56c41e5da8922ae8","_number":694212,"virtual_id_number":694212,"owner":{"_account_id":19184,"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","username":"idan-hefetz"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-18 11:10:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"all":[{"value":2,"date":"2019-11-18 09:58:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-14 16:16:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"all":[{"value":1,"date":"2019-11-18 09:58:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-14 16:16:26.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-11-18 09:58:30.000000000","updated_by":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"reviewer":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"state":"REVIEWER"},{"updated":"2019-11-18 11:10:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9372dd68f50b070e635188572286bf84dcff50ae","author":{"_account_id":19184,"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","username":"idan-hefetz"},"date":"2019-11-14 08:39:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f7609ffe0bcc3aeed6ec877e9ab2e776c9bb3e8","author":{"_account_id":19184,"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","username":"idan-hefetz"},"date":"2019-11-14 08:40:15.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 47cf3434912dcf0e94ebf2a00597cf6da5736727","accounts_in_message":[],"_revision_number":1},{"id":"3dd463dcb4f7e143e07ff890ebe0e49d9a7d1108","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 09:48:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1beffd1d72b4fe0815b45a69602f770 : FAILURE in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b63ed8cf51d64f02813435b821cfb487 : FAILURE in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf89cd1dcad84306abdf09887c947e68 : FAILURE in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30557f9b51ad45678cef1e78711911bb : FAILURE in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bda94fe47203403f815c51bb0a6e088e : SUCCESS in 4m 24s\n- vitrage-dsvm-api-py3 https://zuul.opendev.org/t/openstack/build/25c082c3189f4d6d89daa4318459485c : SUCCESS in 53m 28s\n- vitrage-dsvm-datasources-py3 https://zuul.opendev.org/t/openstack/build/40f49a67977c47798ea4dc6755aa5a02 : SUCCESS in 1h 03m 26s\n- vitrage-dsvm-mock-py3 https://zuul.opendev.org/t/openstack/build/6335990df9684c9b890e59d7848e23b6 : SUCCESS in 58m 25s (non-voting)\n- grenade-vitrage https://zuul.opendev.org/t/openstack/build/194b412ed2bd4acca812794828e6cbc4 : SUCCESS in 54m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d8b3307017b5ed566f2b30ff60b0aeaef11d9f39","author":{"_account_id":19184,"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","username":"idan-hefetz"},"date":"2019-11-14 13:00:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9a8e3d2f1a1a0749b433ddef9c7e324bf3592945","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 14:13:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d84bf5c0d3b4ea299344562075aa0bf : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f238bf5692a7498a8f5e8c175cf1a49e : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee991f7830d84b58b98837eab3f95b0d : SUCCESS in 6m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/054e115464ee410a94b0a913a5e5bffc : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/275897476b8f4ecb89a6106b9bcd74f0 : SUCCESS in 5m 46s\n- vitrage-dsvm-api-py3 https://zuul.opendev.org/t/openstack/build/f8b37f2a993e42e7af416e099321146f : SUCCESS in 59m 33s\n- vitrage-dsvm-datasources-py3 https://zuul.opendev.org/t/openstack/build/934b0c901e8c44a998540650f9e69d82 : FAILURE in 1h 08m 20s\n- vitrage-dsvm-mock-py3 https://zuul.opendev.org/t/openstack/build/48b5faa438cf4e23aabc911bde06161d : SUCCESS in 53m 53s (non-voting)\n- grenade-vitrage https://zuul.opendev.org/t/openstack/build/414ec397df534423be34a22b880552b4 : SUCCESS in 50m 43s","accounts_in_message":[],"_revision_number":2},{"id":"ae969139806f237080d529d556c620e95177b30c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-14 14:36:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5e2f0b0fa1f039a400c05a2d2547edb1cd5c6130","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 16:11:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2b5697134d74ee782e07691805f2e63 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6329039c83284f64bf6a65cbb1e1ac11 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/beeca36cd6f241ed9f80fddc7aa52c8e : SUCCESS in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ae734c9dffb4b8a95c4037a9788e962 : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09e815cc27704a2caaf49566d062d88c : SUCCESS in 3m 54s\n- vitrage-dsvm-api-py3 https://zuul.opendev.org/t/openstack/build/8768c32ed6fe49a0865a591d2d7dfc57 : SUCCESS in 49m 03s\n- vitrage-dsvm-datasources-py3 https://zuul.opendev.org/t/openstack/build/4920645824634b27b66500df339cbee5 : SUCCESS in 1h 17m 24s\n- vitrage-dsvm-mock-py3 https://zuul.opendev.org/t/openstack/build/2fb73792db3246e0b0f4fd7d45887a85 : SUCCESS in 53m 47s (non-voting)\n- grenade-vitrage https://zuul.opendev.org/t/openstack/build/762b1305af0d414bbf1c3bc3103c2b58 : SUCCESS in 49m 49s","accounts_in_message":[],"_revision_number":2},{"id":"7ed77ff6f4c27e42293149e073cfcd284c7366fc","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-14 16:16:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2d185a7cc954d4b2177d2231ce3ad1cd630763b2","author":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"date":"2019-11-18 09:58:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2cb9d0831e3e56223a457d5c683a769af44fe000","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 09:58:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9b5c4fbc04871b31827e2b7e029bd6c5640ac137","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 11:10:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2ec5bf3158b439c8474ace168a1b69e : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/793fef64530648e68ede608d1213a2b1 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1813747028bc4279a9ac7fe5e83635af : SUCCESS in 7m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c16f9f495214a90ad1096ce057029be : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1da97ae75784097a7670f2254269085 : SUCCESS in 4m 39s\n- vitrage-dsvm-api-py3 https://zuul.opendev.org/t/openstack/build/25c8fc8820d140f487f361116af46122 : SUCCESS in 53m 13s\n- vitrage-dsvm-datasources-py3 https://zuul.opendev.org/t/openstack/build/fde4a24c77fa41beaf7d4d3824ca70e7 : SUCCESS in 1h 07m 31s\n- grenade-vitrage https://zuul.opendev.org/t/openstack/build/65758ed83ba34a819dcd7163fe037bcb : SUCCESS in 52m 38s","accounts_in_message":[],"_revision_number":2},{"id":"dea4241bb56f364b67a0c38993a443c43a10bb8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 11:10:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"10d22861e519d1e2ec2f707138a2c041f07a1645","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 11:12:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5f05b4981f341e09e06bf70105f8877 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e2a792a1c71fa7b1d3d8bbb821943bbf4a882ac6","revisions":{"496ca7b5d699281e0ba5a04f5128fbcb62c12776":{"kind":"REWORK","_number":1,"created":"2019-11-14 08:39:04.000000000","uploader":{"_account_id":19184,"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","username":"idan-hefetz"},"ref":"refs/changes/12/694212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/12/694212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/12/694212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/12/694212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/12/694212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/12/694212/1"}}},"commit":{"parents":[{"commit":"8122f24ebec5d42764673af4df119cd7f583fdaa","subject":"fix tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/8122f24ebec5d42764673af4df119cd7f583fdaa"}]}],"author":{"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","date":"2019-11-14 08:38:18.000000000","tz":0},"committer":{"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","date":"2019-11-14 08:38:18.000000000","tz":0},"subject":"Bugfix - Processor check if neighbors are newer before updating them","message":"Bugfix - Processor check if neighbors are newer before updating them\n\nThe bug is that the processor may update a vertex to an older timestamp\n than the existing.\n\nChange-Id: I67e4c59fc1170b07e8f7c50416de8e8598152f09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/496ca7b5d699281e0ba5a04f5128fbcb62c12776"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/496ca7b5d699281e0ba5a04f5128fbcb62c12776"}]},"branch":"refs/heads/master"},"e2a792a1c71fa7b1d3d8bbb821943bbf4a882ac6":{"kind":"REWORK","_number":2,"created":"2019-11-14 13:00:37.000000000","uploader":{"_account_id":19184,"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","username":"idan-hefetz"},"ref":"refs/changes/12/694212/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/12/694212/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/12/694212/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/12/694212/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/12/694212/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/12/694212/2"}}},"commit":{"parents":[{"commit":"8122f24ebec5d42764673af4df119cd7f583fdaa","subject":"fix tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/8122f24ebec5d42764673af4df119cd7f583fdaa"}]}],"author":{"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","date":"2019-11-14 08:38:18.000000000","tz":0},"committer":{"name":"Idan Hefetz","email":"idan.hefetz@nokia.com","date":"2019-11-14 13:00:19.000000000","tz":0},"subject":"Bugfix - Processor check if neighbors are newer before updating them","message":"Bugfix - Processor check if neighbors are newer before updating them\n\nThe bug is that the processor may update a vertex to an older timestamp\n than the existing.\n\nChange-Id: I67e4c59fc1170b07e8f7c50416de8e8598152f09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/e2a792a1c71fa7b1d3d8bbb821943bbf4a882ac6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/e2a792a1c71fa7b1d3d8bbb821943bbf4a882ac6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
