)]}'
{"id":"openstack%2Fvitrage~695138","triplet_id":"openstack%2Fvitrage~master~I1a7044ea10b94cdb25dfcc05d3b2b77d877c1dd5","project":"openstack/vitrage","branch":"master","hashtags":[],"change_id":"I1a7044ea10b94cdb25dfcc05d3b2b77d877c1dd5","subject":"Start README.rst with a better title","status":"MERGED","created":"2019-11-20 07:27:51.000000000","updated":"2019-12-10 14:57:20.000000000","submitted":"2019-12-10 14:55:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695138-1575989741136-aff7a9a8","meta_rev_id":"bba6b7e05202d7941e0fd971e14449f4cc3cb899","_number":695138,"virtual_id_number":695138,"owner":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-10 14:55:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":2,"date":"2019-12-10 14:43:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-27 09:49:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":1,"date":"2019-12-10 14:43:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-27 09:49:01.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2019-12-10 14:43:25.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-12-10 14:55:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"649f9442816481f1a37e5370571a8800bb4c8111","author":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"date":"2019-11-20 07:27:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"131c070f9460092dd31ecc369287c3e0739590ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 07:33:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/42349710314e467e8a946e188c756c16 : SUCCESS in 2m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c13e52d3b89a454597f29d38165fdf6b : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d56f59ab81db461e8c78636656a27d15 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"ec482e23ffae801564b4e69a2003ead211cf2514","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-11-27 09:49:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3cda96b8e164cb4d7712ed1402a2f0171884c7fa","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-12-10 14:43:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1d92d77f2f2e190c6183a9ba241c511287e48fe3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 14:43:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"90586a68ee47ac71fd0528bf7c7a22f68dbb3160","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 14:55:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7b93e3a99cfe4ae2ad501ca17bdd6d22 : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee3d57a263484121b7e74803327f4095 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d5df2963c3846e8b0f9c04049df881b : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"349cf3ce443cc694b3beaae72ef6d354fdf4b272","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 14:55:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bba6b7e05202d7941e0fd971e14449f4cc3cb899","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 14:57:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f1b8d6010814b82a0bde695e9a74916 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0b16f2e933b7bc9c07726c6ebff03b0a0e1226f2","revisions":{"0b16f2e933b7bc9c07726c6ebff03b0a0e1226f2":{"kind":"REWORK","_number":1,"created":"2019-11-20 07:27:51.000000000","uploader":{"_account_id":24061,"name":"pangliye","email":"pangliye@inspur.com","username":"pangliye"},"ref":"refs/changes/38/695138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/38/695138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/38/695138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/38/695138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/38/695138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/38/695138/1"}}},"commit":{"parents":[{"commit":"e2a792a1c71fa7b1d3d8bbb821943bbf4a882ac6","subject":"Bugfix - Processor check if neighbors are newer before updating them","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/e2a792a1c71fa7b1d3d8bbb821943bbf4a882ac6"}]}],"author":{"name":"pangliye","email":"pangliye@inspur.com","date":"2019-11-20 07:24:35.000000000","tz":480},"committer":{"name":"pangliye","email":"pangliye@inspur.com","date":"2019-11-20 07:27:09.000000000","tz":480},"subject":"Start README.rst with a better title","message":"Start README.rst with a better title\n\nNow that we are using gitea the contents of our README.rst are\nmore prominently displayed. Starting it with a \"Team and repository\ntags\" title is a bit confusing. This change makes it start with the\nname of the project instead.\n\nWhile at it, make intro more user-friendly and point to CONTRIBUTING.rst\ninstead of HACKING.rst for contributor guidance.\n\nChange-Id: I1a7044ea10b94cdb25dfcc05d3b2b77d877c1dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/0b16f2e933b7bc9c07726c6ebff03b0a0e1226f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/0b16f2e933b7bc9c07726c6ebff03b0a0e1226f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
