)]}'
{"id":"openstack%2Fvitrage~716415","triplet_id":"openstack%2Fvitrage~master~Ic22be897e255bb19f17009fbc38ca282273f3c95","project":"openstack/vitrage","branch":"master","topic":"service-protocol","hashtags":[],"change_id":"Ic22be897e255bb19f17009fbc38ca282273f3c95","subject":"Use SERVICE_PROTOCOL variable in the devstack plugin","status":"MERGED","created":"2020-03-31 20:02:01.000000000","updated":"2020-04-01 07:20:39.000000000","submitted":"2020-04-01 07:18:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716415-1585725516110-73754ddb","meta_rev_id":"1d880ebd66554e45e3cc53c5a03742924dcae15b","_number":716415,"virtual_id_number":716415,"owner":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-01 07:18:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-01 06:00:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-01 06:00:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-01 06:00:42.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-04-01 07:18:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fda3a4ed85c7c5a5cdd74c881b0e571cfe4452f4","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-31 20:02:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e03d0543ea3a83af4e3d6c1ab342e63842174971","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 21:12:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/721215c8eed94ce5afb2ea8b63371f8f : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e06199a7a8e4a0b9f38870093bf1006 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c6973076d8d4629a76415c601c70739 : SUCCESS in 7m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7a57f45d0a445fe8572701e02dd3fb2 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03f1fa8ccece4849ad3f440c90949319 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45d0c30172b44437a336981da96b3c78 : SUCCESS in 4m 43s\n- vitrage-dsvm-api-py3 https://zuul.opendev.org/t/openstack/build/fa48ddb3218041fe95598ca1607b76a3 : SUCCESS in 52m 37s\n- vitrage-dsvm-datasources-py3 https://zuul.opendev.org/t/openstack/build/39ee74d13f5f44398f4a787e32571451 : SUCCESS in 1h 03m 10s\n- vitrage-dsvm-mock-py3 https://zuul.opendev.org/t/openstack/build/25ba4a63ad0b404898946037f44df50f : SUCCESS in 58m 46s (non-voting)\n- grenade-vitrage https://zuul.opendev.org/t/openstack/build/67b0638c73c9418b80fe3a80474dae08 : SUCCESS in 1h 00m 13s\n- vitrage-dsvm-api-ipv6-only https://zuul.opendev.org/t/openstack/build/c5fdbc50f1b74a888cd8eea2b02ce27c : SUCCESS in 53m 00s","accounts_in_message":[],"_revision_number":1},{"id":"09c67260c031290b86fe613d134e4582cc2e3257","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-01 06:00:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"edddca778b8b7e8f95917db3f90553073a090312","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 06:00:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"34946881ea6b9fb5cd056603cf6b6d582f4bc374","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 07:18:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/924ae7fcde8c4ab8965b8a2bab92ef3f : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7686b227c23e46f180589e0e8afa535f : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4f3311b239c4c3d9b3d52923e850bc0 : SUCCESS in 8m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d77c34ba7c0d4ba2bfd857ac56ee89ce : SUCCESS in 8m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5815d308c75546b6b32b2b86cddd6b00 : SUCCESS in 6m 34s\n- vitrage-dsvm-api-py3 https://zuul.opendev.org/t/openstack/build/bb0173be6b244ed6a810f270f2d41637 : SUCCESS in 58m 41s\n- vitrage-dsvm-datasources-py3 https://zuul.opendev.org/t/openstack/build/c946bce381774cfe822e7933ff15437e : SUCCESS in 1h 09m 40s\n- grenade-vitrage https://zuul.opendev.org/t/openstack/build/dcffd067e4c34ac39f911c5fba1914f7 : SUCCESS in 57m 49s\n- vitrage-dsvm-api-ipv6-only https://zuul.opendev.org/t/openstack/build/1513a43061a14cec9d5f599ab24a4321 : SUCCESS in 52m 51s","accounts_in_message":[],"_revision_number":1},{"id":"492001da5755e9e2ee47c05febc0038d4c793ed2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 07:18:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fb04d44b3ec24ebc967032f3f15715dbfc0015b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 07:20:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/153b76bd9de6498b865566e31685266c : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb1bf8619f31657d4c18bba2d351829870d3723f","revisions":{"eb1bf8619f31657d4c18bba2d351829870d3723f":{"kind":"REWORK","_number":1,"created":"2020-03-31 20:02:01.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/15/716415/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/vitrage","ref":"refs/changes/15/716415/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/15/716415/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/15/716415/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/vitrage refs/changes/15/716415/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/vitrage refs/changes/15/716415/1"}}},"commit":{"parents":[{"commit":"35fc976e5ff946d131096fd9e56a6ec799e857f9","subject":"Use the right python plugin for uwsgi","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/35fc976e5ff946d131096fd9e56a6ec799e857f9"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-31 19:56:14.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-31 19:56:14.000000000","tz":180},"subject":"Use SERVICE_PROTOCOL variable in the devstack plugin","message":"Use SERVICE_PROTOCOL variable in the devstack plugin\n\nVitrage service be deployed with TLS in Devstack, so we need to honor\nSERVICE_PROTOCOL variable to set correct value for VITRAGE_SERVICE_PROTOCOL\n\nChange-Id: Ic22be897e255bb19f17009fbc38ca282273f3c95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/eb1bf8619f31657d4c18bba2d351829870d3723f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/vitrage/commit/eb1bf8619f31657d4c18bba2d351829870d3723f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
